)]}'
{"id":"openstack%2Fproject-config~751046","triplet_id":"openstack%2Fproject-config~master~I3c23e1ba909f048ddbfe9d8891b06c8cb2e2c407","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I3c23e1ba909f048ddbfe9d8891b06c8cb2e2c407","subject":"Add some more monasca projects in vexxhost tenant","status":"MERGED","created":"2020-09-10 16:31:51.000000000","updated":"2020-09-10 17:32:46.000000000","submitted":"2020-09-10 17:12:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751046-1599757941547-06ebbb73","meta_rev_id":"7c31c3a813b6dda243c13fa58f436137ecb35302","_number":751046,"virtual_id_number":751046,"owner":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-10 17:12:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-09-10 16:43:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-10 16:56:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-10 16:56:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-10 16:43:13.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-09-10 16:56:02.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-09-10 17:12:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5649105ae63efb09e059f165d82876d0caabcdba","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2020-09-10 16:31:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a56f7a5200b180ca244325d1bc7cf78f4996082","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2020-09-10 16:32:51.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"fe214b021bf797393783b82908c4fcb4f5082a8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 16:41:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/991fb8f18abb4b4f803c774d1a053acc : SUCCESS in 6m 58s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/dc51c8d49a2a443f907b4aa6e03f8c3d : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":2},{"id":"762f850c343437202b320d8c8f552a0c3eb610a4","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-09-10 16:43:13.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5f8ba870021781a1ab137372e3edd68ddd0d0192","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-09-10 16:56:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1a468f1b109f73a9ede51799f49d58659695f61a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 16:56:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"09e14f4d124ad0a751be646312146b06a73a9e53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 17:12:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9882f3fd14fe4fc685b9e33893e77295 : SUCCESS in 7m 38s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/6bb4cd4d9bf3444db14b4c3b64786d11 : SUCCESS in 7m 13s","accounts_in_message":[],"_revision_number":2},{"id":"babbb1b50f515da94f8d7ae48ef5c5c69fb83114","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 17:12:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7c31c3a813b6dda243c13fa58f436137ecb35302","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 17:32:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/069dfc7b4008489fa46dca1b2c9193b2 : SUCCESS in 11m 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"52c139722f8240ffdc1d912b6291ebf90308fc14","revisions":{"004c14b9165031aa7c9b8f1556cc4f9a463b4dab":{"kind":"REWORK","_number":1,"created":"2020-09-10 16:31:51.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/46/751046/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/46/751046/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/46/751046/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/46/751046/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/46/751046/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/46/751046/1"}}},"commit":{"parents":[{"commit":"9f9b47a8421d8e114907395354dd9c9c5c0b854a","subject":"Merge \"pynotedb: remove project from infrastructure systems\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9f9b47a8421d8e114907395354dd9c9c5c0b854a"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-09-10 16:30:33.000000000","tz":180},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-09-10 16:31:47.000000000","tz":180},"subject":"Add some more monasca projects in vexxhost tenant","message":"Add some more monasca projects in vexxhost tenant\n\nChange-Id: I3c23e1ba909f048ddbfe9d8891b06c8cb2e2c407\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/004c14b9165031aa7c9b8f1556cc4f9a463b4dab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/004c14b9165031aa7c9b8f1556cc4f9a463b4dab"}]},"branch":"refs/heads/master"},"52c139722f8240ffdc1d912b6291ebf90308fc14":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-09-10 16:32:51.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/46/751046/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/46/751046/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/46/751046/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/46/751046/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/46/751046/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/46/751046/2"}}},"commit":{"parents":[{"commit":"9f9b47a8421d8e114907395354dd9c9c5c0b854a","subject":"Merge \"pynotedb: remove project from infrastructure systems\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9f9b47a8421d8e114907395354dd9c9c5c0b854a"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-09-10 16:30:33.000000000","tz":180},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-09-10 16:31:58.000000000","tz":180},"subject":"Add some more monasca projects in vexxhost tenant","message":"Add some more monasca projects in vexxhost tenant\n\nNeeded-By: https://review.opendev.org/#/c/750640\nChange-Id: I3c23e1ba909f048ddbfe9d8891b06c8cb2e2c407\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/52c139722f8240ffdc1d912b6291ebf90308fc14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/52c139722f8240ffdc1d912b6291ebf90308fc14"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
