)]}'
{"id":"openstack%2Fproject-config~751141","triplet_id":"openstack%2Fproject-config~master~If2689e266cbb4dcd1ee7b47928b4dee3cc60f969","project":"openstack/project-config","branch":"master","topic":"osops","hashtags":[],"change_id":"If2689e266cbb4dcd1ee7b47928b4dee3cc60f969","subject":"Complete retirement of x/osops-* repos","status":"MERGED","created":"2020-09-11 01:15:50.000000000","updated":"2020-09-30 15:49:21.000000000","submitted":"2020-09-30 15:27:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751141-1601479676738-1d5b8e56","meta_rev_id":"48826ff24c7437e492254a7e3beb84f336339826","_number":751141,"virtual_id_number":751141,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-09-30 13:58:36.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-30 15:27:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-09-30 15:16:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-09-30 14:04:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-09-30 15:16:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-14 14:02:42.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-30 14:04:12.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-09-30 15:16:05.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-09-30 15:27:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cb6a92550ecb646f121c4d9bdab36e6f3feb75de","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-11 01:15:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c26c0994aed04d26931bbebfefe2a84aa39308b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 01:24:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/dfb414f185d1442d8f32a8b211d70da3 : SUCCESS in 5m 53s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/cab512fff94044d4bbac0e11c8b29aac : SUCCESS in 3m 37s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/9f919c7afdad42f4a7dc2c18c6e74609 : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":1},{"id":"9d51c17d37a05c5bb725cd6a3b3fe54126075724","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-14 14:02:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"de57ca0c43e2d2ac482d94750106d70f62b16f73","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-09-28 18:44:39.000000000","message":"Patch Set 1:\n\ncould you help pushing the dependency and rebase this, please?","accounts_in_message":[],"_revision_number":1},{"id":"ce63d299cf2766ecfbb691d7111e37c7b995a059","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-28 18:52:13.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"7fa6f345a1f2e3547dd01520706c38179893dd56","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-09-28 18:58:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"60dc04662d830d3e1bae690b5af04d45bb95a3da","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-09-28 18:58:16.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"172263f34dfab9671d2dbd99b0c56847c269b332","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-28 19:01:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e44019b6ed33354382263265d1eac7e6d681a7ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 19:13:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/eaffc5ed219d4dbf8e6d6964e3dbbb7d : SUCCESS in 5m 21s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b9759e4e91264455a3408d48ccf85313 : SUCCESS in 4m 23s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/1aeb3f1e53184b07ab983b28e63372a4 : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":3},{"id":"e64d87366022d96544569dd012748036087c37f5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-09-28 19:18:20.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"40a928c73eb8a924c12c15c89826edcec6705b53","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-28 19:26:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"595a73606bca32d8f141243a8cca59f5c3b58934","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 19:34:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7a1d6d91bd2145bf9d001e855c0cc2a3 : SUCCESS in 6m 14s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/3e3edfcb68b949fc8e004f0db9ceeac0 : SUCCESS in 4m 28s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/7512a335642b45ae811dc30ad14a5a07 : SUCCESS in 7m 03s","accounts_in_message":[],"_revision_number":4},{"id":"f19f8ad4595aa6a770097d20a7c950308c289e00","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-09-29 06:29:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"038abc6ae854a265f700059121c98650192a2a8c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-09-29 06:30:20.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"d98b4d3b5b261315dbd1c3886f46a0e4d3706b5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 06:44:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/bfc1b6afccdd4e65a9cc2514a021ec15 : SUCCESS in 6m 48s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/9a5e8ab7f5af48d9b918d3f4e39042cd : SUCCESS in 4m 10s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/9b33ce439b6c4cd5b427784c6de371bc : SUCCESS in 8m 08s","accounts_in_message":[],"_revision_number":5},{"id":"3c8e38a4d7389ff4de8f2767f104cf00457b8b5d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-09-30 11:14:59.000000000","message":"Patch Set 5: Workflow-1\n\nThere\u0027re still a few changes not abandoned, please abandon first:\n\nhttps://review.opendev.org/#/q/projects:x/osops+is:open\n\nAlso, we prefer README.rst for the final note and not other names. For x, I think it\u0027s a recommendation - for openstack mandatory. Do you want to change or keep as is?","accounts_in_message":[],"_revision_number":5},{"id":"29d79b76a1a496be3b8d29c38c6945fad3f3d50d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-30 13:58:36.000000000","message":"Patch Set 5:\n\n\u003e There\u0027re still a few changes not abandoned, please abandon first:\n \u003e \n \u003e https://review.opendev.org/#/q/projects:x/osops+is:open\n \u003e \nAll should be abandoned now.\n\n \u003e Also, we prefer README.rst for the final note and not other names.\n \u003e For x, I think it\u0027s a recommendation - for openstack mandatory. Do\n \u003e you want to change or keep as is?\n\nLet\u0027s just keep as is to avoid needing to do any more work with those repos and so we can keep it with the convention of just checking out HEAD~1 to get back to the former content.","accounts_in_message":[],"_revision_number":5},{"id":"9d095cf92dedff8385bc523a77ff4ab9fc4c935b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-09-30 14:04:12.000000000","message":"Patch Set 5: Code-Review+2 -Workflow","accounts_in_message":[],"_revision_number":5},{"id":"baddaf26f77c44d410536399a69a4a0a9c35ea5b","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-09-30 15:16:05.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"3168b7c2b250b2fc3c01860efd4c010cd045471f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 15:18:56.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"33d34ddb01fe1cc78fa78d69362a24505cde8572","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 15:27:56.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/61bc17a34a63469f9740029ba3756dd8 : SUCCESS in 5m 36s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f64cfcde53d947d385eddf9d05cf7639 : SUCCESS in 4m 56s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/4ff738c9eb224cd79c932e52f06cde17 : SUCCESS in 7m 40s","accounts_in_message":[],"_revision_number":5},{"id":"8356b946ec0b409fb04912c4e723764ffbea9f7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 15:27:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"4b4b5f30f67da819f63e69e672019a97783e0231","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 15:49:21.000000000","message":"Patch Set 5:\n\nBuild failed (deploy pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/215041f25ff14cd89a224fb97166eb21 : FAILURE in 8m 17s\n- infra-prod-service-eavesdrop https://zuul.opendev.org/t/openstack/build/17613b6eb4f94c1eaf5d6db3ac437955 : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"99fb228f1930f14dc1b4bed0c100a606494b6906","revisions":{"dd4bcddd9b0d1ff74d86cb204bcbf13a0e41e1d8":{"kind":"REWORK","_number":1,"created":"2020-09-11 01:15:50.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/41/751141/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/41/751141/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/751141/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/751141/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/751141/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/41/751141/1"}}},"commit":{"parents":[{"commit":"a88f8a4cadcc5e9761f48a52255c065dc6538942","subject":"Move x/osops-* to openstack/osops","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a88f8a4cadcc5e9761f48a52255c065dc6538942"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-09-11 01:15:00.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-09-11 01:15:00.000000000","tz":-300},"subject":"Complete retirement of x/osops-* repos","message":"Complete retirement of x/osops-* repos\n\nDepends-on: https://review.opendev.org/751138\n\nChange-Id: If2689e266cbb4dcd1ee7b47928b4dee3cc60f969\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dd4bcddd9b0d1ff74d86cb204bcbf13a0e41e1d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dd4bcddd9b0d1ff74d86cb204bcbf13a0e41e1d8"}]},"branch":"refs/heads/master"},"88b075a82e9a19f8ebd256fca8a3fc3ca38cb985":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-09-28 18:52:13.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/41/751141/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/41/751141/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/751141/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/751141/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/751141/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/41/751141/2"}}},"commit":{"parents":[{"commit":"a92ed9a6665486e0438e13af458184c69e2d5f9a","subject":"Move gerritlib jobs to in-repo config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a92ed9a6665486e0438e13af458184c69e2d5f9a"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-09-11 01:15:00.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-09-28 18:52:05.000000000","tz":-300},"subject":"Complete retirement of x/osops-* repos","message":"Complete retirement of x/osops-* repos\n\nDepends-on: https://review.opendev.org/751138\n\nChange-Id: If2689e266cbb4dcd1ee7b47928b4dee3cc60f969\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/88b075a82e9a19f8ebd256fca8a3fc3ca38cb985"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/88b075a82e9a19f8ebd256fca8a3fc3ca38cb985"}]},"branch":"refs/heads/master"},"d2311235155e6dc9d2208de33c28b8205911b620":{"kind":"REWORK","_number":3,"created":"2020-09-28 19:01:03.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/41/751141/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/41/751141/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/751141/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/751141/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/751141/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/41/751141/3"}}},"commit":{"parents":[{"commit":"a92ed9a6665486e0438e13af458184c69e2d5f9a","subject":"Move gerritlib jobs to in-repo config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a92ed9a6665486e0438e13af458184c69e2d5f9a"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-09-11 01:15:00.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-09-28 19:00:55.000000000","tz":-300},"subject":"Complete retirement of x/osops-* repos","message":"Complete retirement of x/osops-* repos\n\nDepends-on: https://review.opendev.org/751138\n\nChange-Id: If2689e266cbb4dcd1ee7b47928b4dee3cc60f969\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d2311235155e6dc9d2208de33c28b8205911b620"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d2311235155e6dc9d2208de33c28b8205911b620"}]},"branch":"refs/heads/master"},"75a4095bfe8ec21a8b13dd288a6a988866b93436":{"kind":"REWORK","_number":4,"created":"2020-09-28 19:26:41.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/41/751141/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/41/751141/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/751141/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/751141/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/751141/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/41/751141/4"}}},"commit":{"parents":[{"commit":"a92ed9a6665486e0438e13af458184c69e2d5f9a","subject":"Move gerritlib jobs to in-repo config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a92ed9a6665486e0438e13af458184c69e2d5f9a"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-09-11 01:15:00.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-09-28 19:26:34.000000000","tz":-300},"subject":"Complete retirement of x/osops-* repos","message":"Complete retirement of x/osops-* repos\n\nDepends-on: https://review.opendev.org/751138\n\nChange-Id: If2689e266cbb4dcd1ee7b47928b4dee3cc60f969\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/75a4095bfe8ec21a8b13dd288a6a988866b93436"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/75a4095bfe8ec21a8b13dd288a6a988866b93436"}]},"branch":"refs/heads/master"},"99fb228f1930f14dc1b4bed0c100a606494b6906":{"kind":"REWORK","_number":5,"created":"2020-09-29 06:30:20.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/41/751141/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/41/751141/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/751141/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/751141/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/751141/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/41/751141/5"}}},"commit":{"parents":[{"commit":"a92ed9a6665486e0438e13af458184c69e2d5f9a","subject":"Move gerritlib jobs to in-repo config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a92ed9a6665486e0438e13af458184c69e2d5f9a"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-09-11 01:15:00.000000000","tz":-300},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2020-09-29 06:30:15.000000000","tz":0},"subject":"Complete retirement of x/osops-* repos","message":"Complete retirement of x/osops-* repos\n\nDepends-on: https://review.opendev.org/751138\n\nChange-Id: If2689e266cbb4dcd1ee7b47928b4dee3cc60f969\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/99fb228f1930f14dc1b4bed0c100a606494b6906"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/99fb228f1930f14dc1b4bed0c100a606494b6906"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
