)]}'
{"id":"openstack%2Fproject-config~752540","triplet_id":"openstack%2Fproject-config~master~I105b2e34b89fd2b3f13277731e90413d00baf78b","project":"openstack/project-config","branch":"master","topic":"airshipit-gerribot","hashtags":[],"change_id":"I105b2e34b89fd2b3f13277731e90413d00baf78b","subject":"Switch airship\u0027s gerritbot channel","status":"MERGED","created":"2020-09-17 17:50:55.000000000","updated":"2020-09-17 19:23:25.000000000","submitted":"2020-09-17 18:57:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752540-1600369071556-57e98530","meta_rev_id":"30a950d60c8c4040c6c506302f5e5365c9a26253","_number":752540,"virtual_id_number":752540,"owner":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28618,"name":"Drew Walters","email":"drewwalters@microsoft.com","username":"drewwalters96"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-17 18:57:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-09-17 18:37:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-09-17 18:44:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28618,"name":"Drew Walters","email":"drewwalters@microsoft.com","username":"drewwalters96"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2020-09-17 18:44:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28618,"name":"Drew Walters","email":"drewwalters@microsoft.com","username":"drewwalters96"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},{"_account_id":28618,"name":"Drew Walters","email":"drewwalters@microsoft.com","username":"drewwalters96"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-17 17:52:52.000000000","updated_by":{"_account_id":28618,"name":"Drew Walters","email":"drewwalters@microsoft.com","username":"drewwalters96"},"reviewer":{"_account_id":28618,"name":"Drew Walters","email":"drewwalters@microsoft.com","username":"drewwalters96"},"state":"REVIEWER"},{"updated":"2020-09-17 18:37:57.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-09-17 18:44:11.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-09-17 18:57:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8287d978e36b0e104216de4c57fa02f40b5d7215","author":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"date":"2020-09-17 17:50:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b64390ad1784b0dac6b009ed8760937da8838210","author":{"_account_id":28618,"name":"Drew Walters","email":"drewwalters@microsoft.com","username":"drewwalters96"},"date":"2020-09-17 17:52:52.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"95a620345dc99a340787e7a5b6c1008b4f2a2838","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-09-17 17:55:13.000000000","message":"Patch Set 1: Code-Review-1\n\nopenstackinfra is a founder on the channel which is what we want. We should also add this channel to the accessbot config to ensure the other chanops management can be done.\n\nhttps://opendev.org/openstack/project-config/src/branch/master/accessbot/channels.yaml#L106-L107 is where the old channels config is for that (don\u0027t remove the old channel, add the new one).\n\nYou can do that in this change too.","accounts_in_message":[],"_revision_number":1},{"id":"db7d322f7b7a7bcccf55f3f8548931e0b9f725e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 18:00:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/58b301eb7d3e424fb36c1cea704a3603 : SUCCESS in 6m 40s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/504cf07afdaa49a7a2a96d3a76dbc686 : FAILURE in 7m 36s","accounts_in_message":[],"_revision_number":1},{"id":"931c73cf4cced882b46aa1436196cd967ef06863","author":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"date":"2020-09-17 18:27:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bb2748ca2c452d4ce6bb1ac5ace0fddf0b4d47f6","author":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"date":"2020-09-17 18:27:20.000000000","message":"Patch Set 1:\n\nAh, thanks Clark!  Done!","accounts_in_message":[],"_revision_number":1},{"id":"b9aae2b058416aca359988f9c6b10efea3e27d86","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-09-17 18:37:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0e6d59b5284380486193d156bd586e26ece59253","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 18:42:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/07b581d60e404f588b6542277bf9f942 : SUCCESS in 7m 48s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/8c13233948cd46a0a39fbf8559bda039 : SUCCESS in 7m 43s","accounts_in_message":[],"_revision_number":2},{"id":"8d124456391f88677cc3b6faf23ee140727e5bef","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-09-17 18:44:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9d4ad20159c94b8e315fa9752421f1fd73f75f7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 18:44:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"456dbe7a1d2a514f99e493566f714edf9a9eeb8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 18:57:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/42053bb6ddef415ea37c2ca7c20989e5 : SUCCESS in 6m 22s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/073d91808e3f4feea0c2806f13ff5f83 : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":2},{"id":"832beab037d22816c3634ade12b25c54d7b728f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 18:57:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"682496402f9a94de583292fd8796caf219eb6ee9","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 19:23:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-run-accessbot https://zuul.opendev.org/t/openstack/build/366c6daed2894f6080dce15aed349176 : SUCCESS in 9m 30s\n- infra-prod-service-eavesdrop https://zuul.opendev.org/t/openstack/build/4e731e3ff03b47a1896683c64d8abffe : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"53b269356b2afc718bf4898ba81a5579478dc949","revisions":{"df0f7f25f404f684322911a22f217ad82e9e99f3":{"kind":"REWORK","_number":1,"created":"2020-09-17 17:50:55.000000000","uploader":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"ref":"refs/changes/40/752540/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/40/752540/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/752540/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/752540/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/752540/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/40/752540/1"}}},"commit":{"parents":[{"commit":"a76caa9a21c76ac549db7fad094444edce66ce8f","subject":"Cleanup x/devstack-plugin-glusterfs jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a76caa9a21c76ac549db7fad094444edce66ce8f"}]}],"author":{"name":"Matt McEuen","email":"madgin@madgin.net","date":"2020-09-17 17:49:33.000000000","tz":-300},"committer":{"name":"Matt McEuen","email":"madgin@madgin.net","date":"2020-09-17 17:49:33.000000000","tz":-300},"subject":"Switch airship\u0027s gerritbot channel","message":"Switch airship\u0027s gerritbot channel\n\nThis switches the Airship project\u0027s gerritbot config to use the\nnew #airshipit-gerritbot channel, instead of the #airshipit channel.\nThis will help make sure conversation doesn\u0027t get missed in #airshipit.\n\nChange-Id: I105b2e34b89fd2b3f13277731e90413d00baf78b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/df0f7f25f404f684322911a22f217ad82e9e99f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/df0f7f25f404f684322911a22f217ad82e9e99f3"}]},"branch":"refs/heads/master"},"53b269356b2afc718bf4898ba81a5579478dc949":{"kind":"REWORK","_number":2,"created":"2020-09-17 18:27:02.000000000","uploader":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"ref":"refs/changes/40/752540/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/40/752540/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/752540/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/752540/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/752540/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/40/752540/2"}}},"commit":{"parents":[{"commit":"a76caa9a21c76ac549db7fad094444edce66ce8f","subject":"Cleanup x/devstack-plugin-glusterfs jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a76caa9a21c76ac549db7fad094444edce66ce8f"}]}],"author":{"name":"Matt McEuen","email":"madgin@madgin.net","date":"2020-09-17 17:49:33.000000000","tz":-300},"committer":{"name":"Matt McEuen","email":"madgin@madgin.net","date":"2020-09-17 18:26:57.000000000","tz":-300},"subject":"Switch airship\u0027s gerritbot channel","message":"Switch airship\u0027s gerritbot channel\n\nThis switches the Airship project\u0027s gerritbot config to use the\nnew #airshipit-gerritbot channel, instead of the #airshipit channel.\nThis will help make sure conversation doesn\u0027t get missed in #airshipit.\n\nChange-Id: I105b2e34b89fd2b3f13277731e90413d00baf78b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/53b269356b2afc718bf4898ba81a5579478dc949"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/53b269356b2afc718bf4898ba81a5579478dc949"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
