)]}'
{"id":"openstack%2Fproject-config~760190","triplet_id":"openstack%2Fproject-config~master~I0028cac15313a5642f0db51cd82bf1bbd1d2021d","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I0028cac15313a5642f0db51cd82bf1bbd1d2021d","subject":"Create telemetry group and include","status":"MERGED","created":"2020-10-28 17:16:13.000000000","updated":"2020-10-30 13:59:44.000000000","submitted":"2020-10-29 15:20:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760190-1603984849625-9350e9b5","meta_rev_id":"b4aa9a1f79c1e15f601b4b5acbe121d2224018ab","_number":760190,"virtual_id_number":760190,"owner":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-29 15:20:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"recommended":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-10-28 19:38:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-10-29 15:03:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-28 19:56:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-10-29 15:03:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-28 19:38:06.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-10-28 19:56:34.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-10-29 15:03:11.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-10-29 15:20:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ed07768cc41262d27fe27d5e79ab335aaa72d386","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-10-28 17:16:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8dcbeffc46403b97c2704158b15d17883c95e974","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-10-28 17:19:03.000000000","message":"Patch Set 1:\n\nYou might also consider adding or replacing a telemetry group for storyboard in the projects.yaml (instead of or in addition to the ceilometer group they already share).","accounts_in_message":[],"_revision_number":1},{"id":"11306439ca12b98f46273fe24d16bd088491ee74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 17:26:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f3b25571248e4115b722a2a19920761c : SUCCESS in 6m 11s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/3beed2cbb2914664ad6e0d5cb336e9f9 : FAILURE in 3m 21s","accounts_in_message":[],"_revision_number":1},{"id":"055652b67f140b53f93e00e540f41200af4a8b04","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-28 17:45:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"37ac0cdad839fbe0259ac8138d7a8401c6cb5d0a","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-10-28 18:52:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5af7ad593137c66b059d4ffc5e5ac4a76d688e67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 19:03:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/749aa3e48f6147fa9325a41566c0af77 : SUCCESS in 5m 23s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/de6c5d8287ae46ef93e1b96ec5de3af1 : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":2},{"id":"96c99c6f7e49b762a39956b709cf30cecb98607d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-28 19:38:06.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b46c681b5e66918608d513b0adf967b7d46727d1","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-10-28 19:56:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1fbbf85f306f87c08c6463a03d7bc57daeb013d2","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-10-29 11:00:22.000000000","message":"Patch Set 2:\n\nJeremy, can we rename the storyboard group easily this way?","accounts_in_message":[],"_revision_number":2},{"id":"7c27bbeaabc782e7471da4acfea43cec66323a69","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-10-29 15:02:11.000000000","message":"Patch Set 2:\n\nYes, I believe the old ceilometer group will just be removed by our jeepyb:manage-projects automation, but worst case I can remove the database record(s) if it doesn\u0027t.","accounts_in_message":[],"_revision_number":2},{"id":"ef27923e6d7a7ab9f15a0e5a4e82a7e6fd1120c0","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-10-29 15:03:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nOk, let\u0027s try ;)","accounts_in_message":[],"_revision_number":2},{"id":"78a2b48f54dda2d2783796320d5eafc9bb7c3993","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 15:03:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0408704f899278b822373b95f3ca6035d12d45ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 15:20:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/dc938eb017d245238f178d9f4df58042 : SUCCESS in 6m 00s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/5d27c544a9bb44d4994eb5593c2862aa : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":2},{"id":"8e29dad9ffb29aed6829af7f3773a4cddf66bd72","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 15:20:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"5fff9cc981f8a7bce3570e73dac7ee67fa630de1","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 15:29:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/b6204ed960064eeb99e7d4b65d0c9803 : SUCCESS in 7m 42s","accounts_in_message":[],"_revision_number":2},{"id":"5b9ae921c684ec3294de8d83183af07580d4abf3","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-10-29 23:32:17.000000000","message":"Patch Set 2:\n\nYep, worked as expected. https://storyboard.openstack.org/#!/project_group/telemetry now exists and https://storyboard.openstack.org/#!/project_group/ceilometer is no longer there.","accounts_in_message":[],"_revision_number":2},{"id":"9a4db648ff45cd4e0b0e999c8c1ae4b61e9b2fcc","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-10-30 10:24:37.000000000","message":"Patch Set 2:\n\nThank you.\n\nCan we please add the folks from ceilometer-core to telemetry-core and to telemetry-stable-core?","accounts_in_message":[],"_revision_number":2},{"id":"b4aa9a1f79c1e15f601b4b5acbe121d2224018ab","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-10-30 13:59:44.000000000","message":"Patch Set 2:\n\nAs discussed in IRC, I have added you to these groups now and you should be able to add anyone else you need.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"305b58683e705c999b51c495506dc24e746ae041","revisions":{"ab20752a034cf63f8ab7a9ec3e2739a55e1a463c":{"kind":"REWORK","_number":1,"created":"2020-10-28 17:16:13.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/90/760190/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/90/760190/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/760190/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/760190/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/760190/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/90/760190/1"}}},"commit":{"parents":[{"commit":"ecbab17e1643b4c49d4b95dddabe7489d8abcda6","subject":"Merge \"Add publish-to-pypi template to networking-l2gw\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ecbab17e1643b4c49d4b95dddabe7489d8abcda6"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2020-10-28 16:33:19.000000000","tz":60},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2020-10-28 17:16:11.000000000","tz":60},"subject":"Create telemetry group and include","message":"Create telemetry group and include\n\naodh\nceilometer\npanko\ntelemetry-specs\n\nChange-Id: I0028cac15313a5642f0db51cd82bf1bbd1d2021d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ab20752a034cf63f8ab7a9ec3e2739a55e1a463c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ab20752a034cf63f8ab7a9ec3e2739a55e1a463c"}]},"branch":"refs/heads/master"},"305b58683e705c999b51c495506dc24e746ae041":{"kind":"REWORK","_number":2,"created":"2020-10-28 18:52:22.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/90/760190/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/90/760190/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/760190/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/760190/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/760190/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/90/760190/2"}}},"commit":{"parents":[{"commit":"ecbab17e1643b4c49d4b95dddabe7489d8abcda6","subject":"Merge \"Add publish-to-pypi template to networking-l2gw\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ecbab17e1643b4c49d4b95dddabe7489d8abcda6"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2020-10-28 16:33:19.000000000","tz":60},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2020-10-28 18:51:47.000000000","tz":60},"subject":"Create telemetry group and include","message":"Create telemetry group and include\n\naodh\nceilometer\npanko\ntelemetry-specs\n\nAlso rename the storyboard group to telemetry\n\nChange-Id: I0028cac15313a5642f0db51cd82bf1bbd1d2021d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/305b58683e705c999b51c495506dc24e746ae041"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/305b58683e705c999b51c495506dc24e746ae041"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
