)]}'
{"id":"openstack%2Fproject-config~761468","triplet_id":"openstack%2Fproject-config~master~Iaf8a4dac561440cc3337b00a738979db60b014ce","project":"openstack/project-config","branch":"master","topic":"pypa-ci","hashtags":[],"change_id":"Iaf8a4dac561440cc3337b00a738979db60b014ce","subject":"Add pypa tenant","status":"MERGED","created":"2020-11-04 22:01:16.000000000","updated":"2020-11-05 23:06:58.000000000","submitted":"2020-11-05 22:51:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"761468-1604616678703-3f776333","meta_rev_id":"07c3234f84845f508601ef3acd765b47f8df4e3d","_number":761468,"virtual_id_number":761468,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-05 22:51:18.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-11-05 22:19:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-05 22:41:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-05 22:19:26.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-11-05 22:51:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fb709ea51c3b6158a90cc0a83632f5a546b3776d","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-11-04 22:01:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e359ebc045f666e07a08343e9057dc0835ba27ce","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-11-04 22:03:00.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"54cfb1da41adae5ff87d67f1ae1d181599501eaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 22:12:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5aebd5ef26fe437eb580a9999761a5ea : FAILURE in 4m 54s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/9662e3b8b5e2401c819a64074c3a6a84 : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":1},{"id":"464816628beffe23296af212e12443f2abe3c966","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-11-04 23:38:08.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"2ab70f9426f04687da03e492d78e33764eb4e393","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 23:48:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1c0bb2480913459aaf00dd76687f08a6 : SUCCESS in 8m 05s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/ab268781c5c9400c947079e7efc95f8c : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":2},{"id":"5b651fdf0d865517b595074eac913e690cec9277","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-11-05 22:19:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"16ffbb9e6540a0a0373295397a8a99800cb36d34","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-11-05 22:41:06.000000000","message":"Patch Set 2: Workflow+1\n\nadding based on the approval of the pypa project-config; will set this up with no-op jobs in the project-config after merge","accounts_in_message":[],"_revision_number":2},{"id":"befa4949da62aefc009f2779cdf7888a504ade2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 22:41:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cc8643f5c44f0e4f94d1a7d46015a211a0ac1b89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 22:51:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"64d7446c12a85f34507eaab43999c437fb877307","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 22:51:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2760aadd4453495e9087ab32c6cf8cf4 : SUCCESS in 7m 28s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/470ee6d64ac14306922b09c67a32efce : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":2},{"id":"6a38c74d0686299f5498606068ce6c56a330ada7","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-11-05 23:05:19.000000000","message":"Patch Set 2:\n\nRetoactive +2 from me.","accounts_in_message":[],"_revision_number":2},{"id":"1d7d51777c028a4bec934b632952eebcd18bc03d","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 23:06:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/4c16380a610146d49ba248d1b079a8aa : SUCCESS in 10m 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8732e1c67c935e12eb0a1d32b291c1a301dba5a3","revisions":{"ba591a9d1b1c41d0b03a51a3180d5b362c35b9e9":{"kind":"REWORK","_number":1,"created":"2020-11-04 22:01:16.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/68/761468/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/68/761468/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/761468/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/761468/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/761468/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/68/761468/1"}}},"commit":{"parents":[{"commit":"2b754d393f011f541095d3b35f6a0cc8a6c06a9b","subject":"Add pypa/project-config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2b754d393f011f541095d3b35f6a0cc8a6c06a9b"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-11-04 21:53:50.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-11-04 22:00:25.000000000","tz":660},"subject":"Add pypa tenant","message":"Add pypa tenant\n\nThis is a tenant for running 3rd-party CI against the pypa projects on\nGithub, particularly pip.\n\nChange-Id: Iaf8a4dac561440cc3337b00a738979db60b014ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ba591a9d1b1c41d0b03a51a3180d5b362c35b9e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ba591a9d1b1c41d0b03a51a3180d5b362c35b9e9"}]},"branch":"refs/heads/master"},"8732e1c67c935e12eb0a1d32b291c1a301dba5a3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-11-04 23:38:08.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/68/761468/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/68/761468/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/761468/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/761468/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/761468/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/68/761468/2"}}},"commit":{"parents":[{"commit":"dc5a86dbe55af1f9f63839e202003523195760fc","subject":"Add pypa/project-config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dc5a86dbe55af1f9f63839e202003523195760fc"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-11-04 21:53:50.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-11-04 23:37:51.000000000","tz":660},"subject":"Add pypa tenant","message":"Add pypa tenant\n\nThis is a tenant for running 3rd-party CI against the pypa projects on\nGithub, particularly pip.\n\nChange-Id: Iaf8a4dac561440cc3337b00a738979db60b014ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8732e1c67c935e12eb0a1d32b291c1a301dba5a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8732e1c67c935e12eb0a1d32b291c1a301dba5a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
