)]}'
{"id":"openstack%2Fproject-config~762205","triplet_id":"openstack%2Fproject-config~master~Ib974111a22e669b5ee24618c5713dbbd11d85371","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"Ib974111a22e669b5ee24618c5713dbbd11d85371","subject":"Add openSUSE Leap 15.1 for stein","status":"ABANDONED","created":"2020-11-10 20:19:07.000000000","updated":"2020-11-11 19:13:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7e970083ebbf341d86da163097230a492d6cf59a","_number":762205,"virtual_id_number":762205,"owner":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"date":"2020-11-10 22:33:45.000000000","_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-11-10 20:42:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":-1,"date":"2020-11-10 22:16:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-10 20:42:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-10 22:16:12.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"22948572f0a7a60a1faf1774a2ca35c42c8f15e8","author":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"date":"2020-11-10 20:19:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c122ddfa9bb8db599ebfc115619fd2367ad1689c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 20:42:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e2e42729420b4fe5a001bec14e31db5b : SUCCESS in 6m 27s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/160a5fcbd60a4650a69e35d78b994734 : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":1},{"id":"4ad02ffadfa7de244e9716eac26a86c0579c9a9c","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-11-10 22:16:12.000000000","message":"Patch Set 1: Code-Review-1\n\nA while back we decided we would only target latest opensuse leap 15. The reason for this is the older versions drop out of support and keeping track of that many extra images is difficult. The switch from 15.1 to 15.2 on the opensuse-15 label is the result of this.\n\n15.1 will only be supported until January. I think it is better to either update the software to continue to support latest opensuse 15 or stop running those tests.","accounts_in_message":[],"_revision_number":1},{"id":"eb3405f40fa900a178c6bc6aa0d2ea5b8f3eb107","author":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"date":"2020-11-10 22:33:45.000000000","message":"Patch Set 1:\n\nUploaded https://review.opendev.org/#/c/762214/ as an alternative solution.","accounts_in_message":[],"_revision_number":1},{"id":"7e970083ebbf341d86da163097230a492d6cf59a","author":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"date":"2020-11-11 19:13:29.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f3b2577a383ea05a13a97847fff84c5e03894e21","revisions":{"f3b2577a383ea05a13a97847fff84c5e03894e21":{"kind":"REWORK","_number":1,"created":"2020-11-10 20:19:07.000000000","uploader":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"ref":"refs/changes/05/762205/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/05/762205/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/762205/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/762205/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/762205/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/05/762205/1"}}},"commit":{"parents":[{"commit":"d1f976cb43347ef76eb584b228644678b906cc23","subject":"Rename testinfra project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d1f976cb43347ef76eb584b228644678b906cc23"}]}],"author":{"name":"Jan Zerebecki","email":"jzerebecki@suse.com","date":"2020-11-10 20:16:17.000000000","tz":60},"committer":{"name":"Jan Zerebecki","email":"jzerebecki@suse.com","date":"2020-11-10 20:18:54.000000000","tz":60},"subject":"Add openSUSE Leap 15.1 for stein","message":"Add openSUSE Leap 15.1 for stein\n\nIn 7a1722900b689b83c8736b89b629ccc7ca5bfd9f opensuse-15 was changed from\n15.1 to 15.2, but 15.2 has libvirt 6.0.0 and stein upper-constraints is\nonly compatible with 5.1.0.\n\nSee https://review.opendev.org/#/c/762204/\naka I09dffabc1aefac412a83c2a202c42288952e6ea3 .\n\nChange-Id: Ib974111a22e669b5ee24618c5713dbbd11d85371\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f3b2577a383ea05a13a97847fff84c5e03894e21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f3b2577a383ea05a13a97847fff84c5e03894e21"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
