)]}'
{"id":"openstack%2Fproject-config~762710","triplet_id":"openstack%2Fproject-config~master~I5444c7633f880c050284b0395ec01a3abf0592f6","project":"openstack/project-config","branch":"master","topic":"pypa-ci","hashtags":[],"change_id":"I5444c7633f880c050284b0395ec01a3abf0592f6","subject":"Correct version range of ansible with ansible-lint","status":"ABANDONED","created":"2020-11-13 19:15:09.000000000","updated":"2020-11-14 18:48:19.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2e0d4f2a6f755a75dee7e3dfc95f0676fd3aaead","_number":762710,"virtual_id_number":762710,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-11-13 19:33:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":-1,"date":"2020-11-14 09:28:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":2,"date":"2020-11-13 19:20:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-13 19:20:39.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-11-13 19:33:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-14 09:28:47.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"state":"REVIEWER"}],"messages":[{"id":"1e92086a3811d62e7e2de8edbcd457ba3ffa5469","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-11-13 19:15:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08e1155e9c71e703584f793861d351fbc48d986c","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-11-13 19:20:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"49c905cfb99552bd282ed0fc0c9893d7a515b46e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-13 19:33:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/4e075a576aa24e68bbbe0dd5a9cd2515 : SUCCESS in 5m 41s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4bd8937fbea14c1fb238b1a479b266c9 : SUCCESS in 5m 39s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/c3fe6722c3c24685a4ad67ac237addef : SUCCESS in 3m 15s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/99554ecb38824732b35810b4ba317026 : SUCCESS in 3m 42s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/f9dac4bd7e834477a9b76998c1c52071 : SUCCESS in 3m 21s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/37664f21817348919a01af9a48a6efc8 : SUCCESS in 5m 03s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/bc3b907d5fbc4d45a437fd82c92c3325 : SUCCESS in 6m 55s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/0acbe1fd1dd2408a985e89e4f4cb958a : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":1},{"id":"7523e54035ffb8226ffb58f35113b97ec22d2adf","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-11-14 09:28:47.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1bc7c7d56cb5cc2ed96f6a5e351e7cdb6830f7b6","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-11-14 18:48:19.000000000","message":"Abandoned\n\nThe point of this change is that a version range should end on an exclusive inequality for its upper bound, because inclusive inequalities have potential side effects. But it sounds like we\u0027ll just need to rip all this out when the next version of pip comes out, so I don\u0027t feel like arguing the point.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"10942587e532825acd6c5d5711ae8ff6353abf5b","revisions":{"10942587e532825acd6c5d5711ae8ff6353abf5b":{"kind":"REWORK","_number":1,"created":"2020-11-13 19:15:09.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/10/762710/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/10/762710/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/762710/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/762710/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/762710/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/10/762710/1"}}},"commit":{"parents":[{"commit":"d7d74c283238191b6709dda5e7b69cf317f7ca6e","subject":"Create inspur/venus-tempest-plugin repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d7d74c283238191b6709dda5e7b69cf317f7ca6e"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2020-11-13 19:13:24.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2020-11-13 19:13:24.000000000","tz":0},"subject":"Correct version range of ansible with ansible-lint","message":"Correct version range of ansible with ansible-lint\n\nUse an exclusive \"strictly less than\" inequality when defining the\nupper bound for the ansible package in the linters testenv deps\nlist, so that ansible 2.9 will not be chosen.\n\nChange-Id: I5444c7633f880c050284b0395ec01a3abf0592f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/10942587e532825acd6c5d5711ae8ff6353abf5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/10942587e532825acd6c5d5711ae8ff6353abf5b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
