)]}'
{"id":"openstack%2Fproject-config~762819","triplet_id":"openstack%2Fproject-config~master~I6c8d9a7eaff4003afbe14b8dfaa58f59f619fa9f","project":"openstack/project-config","branch":"master","topic":"add-magpie","hashtags":[],"change_id":"I6c8d9a7eaff4003afbe14b8dfaa58f59f619fa9f","subject":"Add Magpie charm to OpenStack charms","status":"MERGED","created":"2020-11-16 10:59:37.000000000","updated":"2020-11-20 08:18:04.000000000","submitted":"2020-11-20 06:29:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"762819-1605853746239-11ac1cc1","meta_rev_id":"49c80d59b1a8b979d9c75280e02dcc998011065c","_number":762819,"virtual_id_number":762819,"owner":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-20 06:29:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-11-16 19:05:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-11-16 11:19:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-20 06:09:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-20 06:09:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-16 11:19:26.000000000","updated_by":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"reviewer":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"state":"REVIEWER"},{"updated":"2020-11-16 19:05:02.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-11-20 06:09:55.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-11-20 06:29:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4492a7aa3195361402aa4481e484f0ceb05aae56","author":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"date":"2020-11-16 10:59:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ffce8d4e56ef565ac6c680457fcd68f2786eb91","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2020-11-16 11:19:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"21c60af13f10ffb6a4fe02481ca7dd8bbdd57e42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 11:22:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/85a7cd2b46b346209f3cae37ec995392 : SUCCESS in 6m 00s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/4c411273eaad4aef9a5e5a3b2c15a62c : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"fe8732eab17cfc779aa5fa19e8878896c1b656aa","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-11-16 19:05:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"eef89caf33cf68a6576ecdc0dafe7b077658c8c5","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-11-20 06:09:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0cf7efd58bb11dd3b7b5939144c06bfdf74d4aaa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 06:10:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8a1f6850a52f1f0f1d53b34753a031eba4dff0c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 06:29:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/032b07559113492e9659a541781c9de5 : SUCCESS in 13m 15s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b844bf45912444dbba8cbcfea5daba75 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":1},{"id":"a1794c915af2a809ffe03638962e43d136b71ebe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 06:29:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b161ba107d1b65d438258c918e3b5acfd1d84304","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 08:18:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/6c1b89b7637045fb9011f72b9884886e : SUCCESS in 4m 36s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/34049dc28fb5453a9a003aa666a9a25e : SUCCESS in 11m 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"779ac065b309d2c686d3ea0007aea07fdebe0513","revisions":{"779ac065b309d2c686d3ea0007aea07fdebe0513":{"kind":"REWORK","_number":1,"created":"2020-11-16 10:59:37.000000000","uploader":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"ref":"refs/changes/19/762819/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/19/762819/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/762819/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/762819/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/762819/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/19/762819/1"}}},"commit":{"parents":[{"commit":"d7d74c283238191b6709dda5e7b69cf317f7ca6e","subject":"Create inspur/venus-tempest-plugin repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d7d74c283238191b6709dda5e7b69cf317f7ca6e"}]}],"author":{"name":"Liam Young","email":"liam.young@canonical.com","date":"2020-11-16 10:56:20.000000000","tz":0},"committer":{"name":"Liam Young","email":"liam.young@canonical.com","date":"2020-11-16 10:59:05.000000000","tz":0},"subject":"Add Magpie charm to OpenStack charms","message":"Add Magpie charm to OpenStack charms\n\nAdd Magpie charm and corresponding interface to OpenStack Charms.\n\nChange-Id: I6c8d9a7eaff4003afbe14b8dfaa58f59f619fa9f\nRequired-By: Ia0482284f1d1a18f65437c6d2391a363a7ed8f18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/779ac065b309d2c686d3ea0007aea07fdebe0513"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/779ac065b309d2c686d3ea0007aea07fdebe0513"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
