)]}'
{"id":"openstack%2Fproject-config~763808","triplet_id":"openstack%2Fproject-config~master~Id7976d573b7eeda3ccf5970104e22a4427406e87","project":"openstack/project-config","branch":"master","topic":"cc","hashtags":["git-review"],"change_id":"Id7976d573b7eeda3ccf5970104e22a4427406e87","subject":"Move git-review zuul config in-tree","status":"MERGED","created":"2020-11-23 17:08:37.000000000","updated":"2021-01-13 10:24:46.000000000","submitted":"2021-01-13 10:24:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"763808-cc","meta_rev_id":"a4502a4145018256f771ecf067de953f2ca8d6dc","_number":763808,"virtual_id_number":763808,"owner":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-13 10:24:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2021-01-13 08:37:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2021-01-13 10:16:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2021-01-13 10:16:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-23 17:09:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-24 09:26:50.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-11-24 09:26:50.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-11-24 09:26:50.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2021-01-13 06:56:06.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"CC"},{"updated":"2021-01-13 08:11:53.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2021-01-13 10:07:19.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2021-01-13 10:16:37.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"238857bd6daf55dce7b7bc5be182e91371bb18e6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-11-23 17:08:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a377f07c727fef915792c3ebfba28b394ab8f0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 17:09:31.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  The project template \"openstack-tox-linters\" was not found.\n\nThe error appears in the following project stanza:\n\n  project:\n      name: opendev/git-review\n      templates:\n        - openstack-tox-linters\n        - publish-to-pypi\n        - publish-tox-docs-infra\n      check:\n        jobs:\n          - tox-py35\n          - tox-py36\n          - tox-py37\n          - tox-py38\n      gate:\n        jobs:\n          - tox-py35\n          - tox-py36\n          - tox-py37\n          - tox-py38\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 129, column 3","accounts_in_message":[],"_revision_number":1},{"id":"ad27388cbc9bd92f745fae3e5eddc4686e6687a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 17:10:11.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  The project template \"openstack-tox-linters\" was not found.\n\nThe error appears in the following project stanza:\n\n  project:\n      name: opendev/git-review\n      templates:\n        - openstack-tox-linters\n        - publish-to-pypi\n        - publish-tox-docs-infra\n      check:\n        jobs:\n          - tox-py35\n          - tox-py36\n          - tox-py37\n          - tox-py38\n      gate:\n        jobs:\n          - tox-py35\n          - tox-py36\n          - tox-py37\n          - tox-py38\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 129, column 3","accounts_in_message":[],"_revision_number":1},{"id":"1dcd25e37aa677124a93421768e91e2ba20eee58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 17:11:01.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  The project template \"openstack-tox-linters\" was not found.\n\nThe error appears in the following project stanza:\n\n  project:\n      name: opendev/git-review\n      templates:\n        - openstack-tox-linters\n        - publish-to-pypi\n        - publish-tox-docs-infra\n      check:\n        jobs:\n          - tox-py35\n          - tox-py36\n          - tox-py37\n          - tox-py38\n      gate:\n        jobs:\n          - tox-py35\n          - tox-py36\n          - tox-py37\n          - tox-py38\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 129, column 3","accounts_in_message":[],"_revision_number":1},{"id":"881c87b1d2f847ba2361f072a0b8958ae8c6a2b4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-11-23 17:14:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7fc136283a8742013c2a08e0cc234fd1eb2f9ef7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 17:26:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9aea75a464d64583b2465326952db8d3 : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":2},{"id":"5b2f478bd2b379b0713b2b95879b5f78d60e4e48","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-11-24 09:25:59.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"ee971ca06f66b223e75af4b43907d5ca2b5a6e1a","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-11-24 09:27:55.000000000","message":"Hashtag added: git-review","accounts_in_message":[],"_revision_number":2},{"id":"23673e59161040ae877933e68122e51843d8b6e8","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-01-12 18:54:24.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a03a5e646e5505e94b1de519ddd05dcd66935218","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2021-01-12 20:16:31.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"a9c2686983cac2bf2ef7cd97756542fa0dc40e14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 20:27:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c72e4d518a664240ba5953618278ad49 : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":3},{"id":"f14fa522f1c5eabb8e40d5609bca4da1fec94a2f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-01-12 20:31:44.000000000","message":"Patch Set 3: Code-Review+2\n\nI\u0027m good with this, though we should move all of this into an in-repo project-pipeline so subsequent changes are self-tested.","accounts_in_message":[],"_revision_number":3},{"id":"3d2d94bec0ad69edda038b17b881786bd3130f7c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-01-13 06:56:06.000000000","message":"Patch Set 3:\n\nCould you move this in-tree instead, please?","accounts_in_message":[],"_revision_number":3},{"id":"89baffce10bbd448934269dc0f76bd0eb2148d8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2021-01-13 08:05:30.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"15073f62e747ebb2bb9bcc0a17f87569d6a405e6","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-01-13 08:11:53.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"b69758bc004b6faf373525881b3a0486027dd705","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2021-01-13 08:21:24.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"cc8dcfb57e239033f0c64432f9f1d8a203583c70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 08:34:54.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/67de67de86624dffa0033f2fa6e8a256 : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":5},{"id":"db410884685d917d265b1ac6355868753aca7cd1","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-01-13 08:37:55.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"cfe6d444b92959449e8944594f9b155c39d732a9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-01-13 10:07:19.000000000","message":"Patch Set 5:\n\nI\u0027m hesitating to approve this while the patch to restore tests in the target repo is still failing. I\u0027m thinking the dependency should be the other way round, first install the in-repo testing, there shouldn\u0027t be a problem when the same tests are applied twice. Then do this one as a second step.","accounts_in_message":[],"_revision_number":5},{"id":"835bdc441699f0c051a726a25be0b8d77d3c363c","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2021-01-13 10:11:41.000000000","message":"Patch Set 5:\n\n\u003e Patch Set 5:\n\u003e \n\u003e I\u0027m hesitating to approve this while the patch to restore tests in the target repo is still failing. I\u0027m thinking the dependency should be the other way round, first install the in-repo testing, there shouldn\u0027t be a problem when the same tests are applied twice. Then do this one as a second step.\n\nRead the comments on the patch, py27 was broken for 3+ months and we have no plans to restore it, especially as this is caused by some unmaintained test deps.\n\nInitial versions of this patch were to just remove support for py27 but this morning Andreas asked\nto just move in-repo, after he saw the previous comments.\n\nThose two activities are in catch-22: dropping py27 and moving jobs to in-tree.","accounts_in_message":[],"_revision_number":5},{"id":"40353614fe620cd44cc6d74595649cb89677cb71","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-01-13 10:16:37.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nhmm, o.k.","accounts_in_message":[],"_revision_number":5},{"id":"f59a19c922c0b83f83cd03c557d6448982a07301","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 10:17:15.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"117e1e7ce81aa2afeb012d8798819cc72c00c798","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 10:24:45.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e34ca97238fc45fabda93669fd4c01f9 : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":5},{"id":"a4502a4145018256f771ecf067de953f2ca8d6dc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 10:24:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f0bfc80e9786775e1ff4ef2e26ca43697326ebc5","revisions":{"bb9a21f8db6a27b6ea34f2e379031bec91f9e883":{"kind":"REWORK","_number":1,"created":"2020-11-23 17:08:37.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/08/763808/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/08/763808/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/763808/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/763808/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/763808/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/08/763808/1"}}},"commit":{"parents":[{"commit":"dec82d5c8f708c7069af0bdf122e8c7e9130b0ed","subject":"Run tag-releases on ubuntu-focal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dec82d5c8f708c7069af0bdf122e8c7e9130b0ed"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2020-11-23 17:08:02.000000000","tz":0},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2020-11-23 17:08:17.000000000","tz":0},"subject":"Update git-review test matrix","message":"Update git-review test matrix\n\nChange-Id: Id7976d573b7eeda3ccf5970104e22a4427406e87\nNeeded-By: https://review.opendev.org/c/opendev/git-review/+/763803\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bb9a21f8db6a27b6ea34f2e379031bec91f9e883"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bb9a21f8db6a27b6ea34f2e379031bec91f9e883"}]},"branch":"refs/heads/master"},"2378ea70ce4f0d8435cfab23f8a88a6a28cfa8d0":{"kind":"REWORK","_number":2,"created":"2020-11-23 17:14:22.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/08/763808/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/08/763808/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/763808/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/763808/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/763808/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/08/763808/2"}}},"commit":{"parents":[{"commit":"dec82d5c8f708c7069af0bdf122e8c7e9130b0ed","subject":"Run tag-releases on ubuntu-focal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dec82d5c8f708c7069af0bdf122e8c7e9130b0ed"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2020-11-23 17:08:02.000000000","tz":0},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2020-11-23 17:14:07.000000000","tz":0},"subject":"Update git-review test matrix","message":"Update git-review test matrix\n\nChange-Id: Id7976d573b7eeda3ccf5970104e22a4427406e87\nNeeded-By: https://review.opendev.org/c/opendev/git-review/+/763803\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2378ea70ce4f0d8435cfab23f8a88a6a28cfa8d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2378ea70ce4f0d8435cfab23f8a88a6a28cfa8d0"}]},"branch":"refs/heads/master"},"d468062d9f5e1b7073431152e4ad9a39cd86b4a9":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-01-12 20:16:31.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/08/763808/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/08/763808/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/763808/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/763808/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/763808/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/08/763808/3"}}},"commit":{"parents":[{"commit":"dec82d5c8f708c7069af0bdf122e8c7e9130b0ed","subject":"Run tag-releases on ubuntu-focal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dec82d5c8f708c7069af0bdf122e8c7e9130b0ed"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2020-11-23 17:08:02.000000000","tz":0},"committer":{"name":"Sorin Sbârnea","email":"ssbarnea@redhat.com","date":"2021-01-12 20:16:12.000000000","tz":0},"subject":"Update git-review test matrix (drop py27)","message":"Update git-review test matrix (drop py27)\n\nThis adds few testing with more modern versions of python while allowing us to drop\nsupport for py27.\n\nChange-Id: Id7976d573b7eeda3ccf5970104e22a4427406e87\nNeeded-By: https://review.opendev.org/c/opendev/git-review/+/763803\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d468062d9f5e1b7073431152e4ad9a39cd86b4a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d468062d9f5e1b7073431152e4ad9a39cd86b4a9"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8ccfe1941db7b90fe5c560cf76a5771b3e9337fe":{"kind":"REWORK","_number":4,"created":"2021-01-13 08:05:30.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/08/763808/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/08/763808/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/763808/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/763808/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/763808/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/08/763808/4"}}},"commit":{"parents":[{"commit":"dec82d5c8f708c7069af0bdf122e8c7e9130b0ed","subject":"Run tag-releases on ubuntu-focal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dec82d5c8f708c7069af0bdf122e8c7e9130b0ed"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2020-11-23 17:08:02.000000000","tz":0},"committer":{"name":"Sorin Sbârnea","email":"ssbarnea@redhat.com","date":"2021-01-13 08:05:19.000000000","tz":0},"subject":"Move git-review zuul config in-tree","message":"Move git-review zuul config in-tree\n\nThis will allow us to change jobs run by zuul on this repo without having to rely on restricted\nproject-config repo.\n\nChange-Id: Id7976d573b7eeda3ccf5970104e22a4427406e87\nNeeded-By: https://review.opendev.org/c/opendev/git-review/+/770539\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8ccfe1941db7b90fe5c560cf76a5771b3e9337fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8ccfe1941db7b90fe5c560cf76a5771b3e9337fe"}]},"branch":"refs/heads/master"},"f0bfc80e9786775e1ff4ef2e26ca43697326ebc5":{"kind":"REWORK","_number":5,"created":"2021-01-13 08:21:24.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/08/763808/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/08/763808/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/763808/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/763808/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/763808/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/08/763808/5"}}},"commit":{"parents":[{"commit":"9401eab93d2917bf6433ced0bdaab4da9de7392e","subject":"Merge \"Create microstack-specs project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9401eab93d2917bf6433ced0bdaab4da9de7392e"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2020-11-23 17:08:02.000000000","tz":0},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2021-01-13 08:21:14.000000000","tz":0},"subject":"Move git-review zuul config in-tree","message":"Move git-review zuul config in-tree\n\nThis will allow us to change jobs run by zuul on this repo without having to rely on restricted\nproject-config repo.\n\nChange-Id: Id7976d573b7eeda3ccf5970104e22a4427406e87\nNeeded-By: https://review.opendev.org/c/opendev/git-review/+/770539\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f0bfc80e9786775e1ff4ef2e26ca43697326ebc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f0bfc80e9786775e1ff4ef2e26ca43697326ebc5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
