)]}'
{"id":"openstack%2Fproject-config~763889","triplet_id":"openstack%2Fproject-config~master~I9facf665936f9ba8529af2457bb97fc5cc6e49b1","project":"openstack/project-config","branch":"master","topic":"new-project","hashtags":[],"change_id":"I9facf665936f9ba8529af2457bb97fc5cc6e49b1","subject":"New Project Request: airship/vino","status":"MERGED","created":"2020-11-23 23:32:56.000000000","updated":"2020-12-10 15:31:57.000000000","submitted":"2020-12-10 14:58:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"763889-new-project","meta_rev_id":"edf026fa5572714b596ea09c7c7fba07c8922211","_number":763889,"virtual_id_number":763889,"owner":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":21991,"name":"Kostyantyn Kalynovskyi","email":"kkalinovskiy@gmail.com","username":"kkalynovskyi"},{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":29131,"name":"diwakar thyagaraj","email":"diwakar.chitoor.thyagaraj@att.com","username":"dt241s@att.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-10 14:58:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-12-10 15:31:57.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-12-10 15:31:57.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":21991,"name":"Kostyantyn Kalynovskyi","email":"kkalinovskiy@gmail.com","username":"kkalynovskyi"},{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":1,"date":"2020-12-08 19:27:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29131,"name":"diwakar thyagaraj","email":"diwakar.chitoor.thyagaraj@att.com","username":"dt241s@att.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-10 14:47:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":21991,"name":"Kostyantyn Kalynovskyi","email":"kkalinovskiy@gmail.com","username":"kkalynovskyi"},{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":29131,"name":"diwakar thyagaraj","email":"diwakar.chitoor.thyagaraj@att.com","username":"dt241s@att.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-10 14:47:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":21991,"name":"Kostyantyn Kalynovskyi","email":"kkalinovskiy@gmail.com","username":"kkalynovskyi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"_account_id":29131,"name":"diwakar thyagaraj","email":"diwakar.chitoor.thyagaraj@att.com","username":"dt241s@att.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-23 23:44:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-25 08:58:40.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-11-25 17:28:24.000000000","updated_by":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"reviewer":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"state":"REVIEWER"},{"updated":"2020-12-01 16:09:25.000000000","updated_by":{"_account_id":21991,"name":"Kostyantyn Kalynovskyi","email":"kkalinovskiy@gmail.com","username":"kkalynovskyi"},"reviewer":{"_account_id":21991,"name":"Kostyantyn Kalynovskyi","email":"kkalinovskiy@gmail.com","username":"kkalynovskyi"},"state":"REVIEWER"},{"updated":"2020-12-08 19:27:09.000000000","updated_by":{"_account_id":29131,"name":"diwakar thyagaraj","email":"diwakar.chitoor.thyagaraj@att.com","username":"dt241s@att.com"},"reviewer":{"_account_id":29131,"name":"diwakar thyagaraj","email":"diwakar.chitoor.thyagaraj@att.com","username":"dt241s@att.com"},"state":"REVIEWER"},{"updated":"2020-12-10 14:47:16.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"d5ad1cb2ec9a43eb34b3ae7da047b139c3945ef7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"date":"2020-11-23 23:32:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1faf7816a6086e11c5214885491a9eb93966203d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 23:44:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/429266e9dfc640a0a77b0fc430c0ab0b : SUCCESS in 9m 42s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/130d9cd593ce4c2eaac7b9c84f802eb9 : FAILURE in 3m 23s","accounts_in_message":[],"_revision_number":1},{"id":"505e0159134fe6270c7e4e5546af650005ee038e","author":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"date":"2020-11-24 19:47:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"33ada2f2509e50de69b95c535e62063a2fe8d181","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 19:58:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/48651fc1667b49bc8f3fac24f19b18d2 : SUCCESS in 7m 08s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8dc1d941f87d4be68f363f95a29242a9 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":1},{"id":"7e3884ed620835c778b4a742e2f52cff5a39129d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-11-25 08:58:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"155c752851f1f423b8a26980d77861dddc91870d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-11-25 08:59:55.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5a3004933ba65fa809164701266f81054676c7fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"date":"2020-11-25 17:05:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b8aa6b50147c0739831d7bd7cf93c7095a04bfa1","author":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"date":"2020-11-25 17:06:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"12ffbeec255a8cd98ddffa4b5ccdaa4be0f06da0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 17:15:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/351363beff8d499bab8925e54736550a : SUCCESS in 6m 51s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/138b9593b04a45c08aa6352057a0e532 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":2},{"id":"f72366b50dd7e59b9419c6c3ed3cddaa24cc84bc","author":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"date":"2020-11-25 17:28:24.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"47d51b011ed0ca25bfff79684174ade1ccffcb9c","author":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"date":"2020-11-25 17:28:41.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"47ba460cc9f8aab03bff5fb4c7db61c7543f9a6f","author":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"date":"2020-11-25 17:29:42.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9365a607c63e666f4c9a459e2a1e7811694647b7","author":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"date":"2020-11-25 17:31:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a500c3cdd353b814afc2810abfc03a81c54bbcf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"date":"2020-11-25 17:33:46.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"b6ac4c951f480ada42943859f197f613a2eaf751","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 17:44:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/dbe83f10972942cb8812ef98e056578e : SUCCESS in 7m 02s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8d49e9c01dcf4065b3a94bb8f76ba511 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":3},{"id":"23052b9e3de01104600614acddd24415e83db8ea","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-11-25 19:33:15.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b7efaaa54ee409bcbc62435da617094ffcd9cfb9","author":{"_account_id":21991,"name":"Kostyantyn Kalynovskyi","email":"kkalinovskiy@gmail.com","username":"kkalynovskyi"},"date":"2020-12-01 16:09:25.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"6f436ff5b3078ebe20aea2d1bf4340631c22476f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"date":"2020-12-01 17:20:33.000000000","message":"Topic new-project-vino removed","accounts_in_message":[],"_revision_number":3},{"id":"1c607112bf551bcf870c17961ff1c30ebffac50d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"date":"2020-12-01 17:20:40.000000000","message":"Topic set to new-project","accounts_in_message":[],"_revision_number":3},{"id":"9e04d837eb74ce11481f77ad321f023af0add681","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"date":"2020-12-07 17:03:12.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"3ec9c83d8f328191ebeb6b43ebb4f9711c0835c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-07 17:14:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/76a8cbaf7634484b9f40931f1e46e1a8 : SUCCESS in 9m 29s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/ecfece94403545479c5ebf9b2b4f2b4c : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":4},{"id":"c862cf430c47dea0fc573d99bd6906adb4b556d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"date":"2020-12-08 16:35:08.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"6c4ffb6122e2671a171eee72bc4ab3a049585687","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 16:48:00.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/bab3c516239f49abadc7ac9f0c2bb04e : SUCCESS in 6m 04s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/dbc635f105fc4773b95f3fbf49eac377 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":5},{"id":"b54f001dc277d7f37163354da85ead9aae9a7d80","author":{"_account_id":29131,"name":"diwakar thyagaraj","email":"diwakar.chitoor.thyagaraj@att.com","username":"dt241s@att.com"},"date":"2020-12-08 19:27:09.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"176d484423012a625a031abb417412f2ba97a57e","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-12-10 14:47:16.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"50b3804054e2f6f7cb049d92dd02fbbe8011f340","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 14:47:37.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"d6b116752447e8ef7cd3cdb8291baff88efa6019","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 14:58:33.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/98abfcedc1c74eaea34740eaf82c09b2 : SUCCESS in 5m 41s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/a000f3dff9ef4dd497460d568c106475 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":5},{"id":"260002580a18fb0c0d47e542bec44990e7780df3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 14:58:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"67c98508eee0ce545f0c9e46a8bd76f7846b19c0","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 15:31:57.000000000","message":"Patch Set 5:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/b688c7f6959e4bee8f97057928af4aa2 : SUCCESS in 7m 42s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/cef97c7dce0d4548a867f9ee9e7f656f : SUCCESS in 9m 59s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"1d3632d9361ca244999b60b4edffef012a2f2b48","revisions":{"feb920cc0fec28ee2ee92d0c1ff2ae9c35ce71c6":{"kind":"REWORK","_number":1,"created":"2020-11-23 23:32:56.000000000","uploader":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"ref":"refs/changes/89/763889/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/89/763889/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/89/763889/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/89/763889/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/89/763889/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/89/763889/1"}}},"commit":{"parents":[{"commit":"f52ca2fe5fe16ab6e3b14237c5ccaaf9a5de64b0","subject":"Add known_hosts entries for additional Gerrit keys","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f52ca2fe5fe16ab6e3b14237c5ccaaf9a5de64b0"}]}],"author":{"name":"Matt McEuen","email":"madgin@madgin.net","date":"2020-11-23 23:22:50.000000000","tz":-360},"committer":{"name":"Matt McEuen","email":"madgin@madgin.net","date":"2020-11-23 23:22:50.000000000","tz":-360},"subject":"New Project Request: airship/vino","message":"New Project Request: airship/vino\n\nThis is a request to create the airship/vino project in opendev.\n\nViNO is based on the Kubernetes ansible operator framework,\nand will have responsibility within Airship-based Kubernetes\nclusters for lifecycle managing static libvirt definitions on\nworker nodes, for use within simple CAPI sub-clusters.\n\nIf you can seed the new gerrit group with me and I can add the appropriate\ninitial core team.\n\nChange-Id: I9facf665936f9ba8529af2457bb97fc5cc6e49b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/feb920cc0fec28ee2ee92d0c1ff2ae9c35ce71c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/feb920cc0fec28ee2ee92d0c1ff2ae9c35ce71c6"}]},"branch":"refs/heads/master"},"2a846c9483597f3398adf8e3909824e3480b6316":{"kind":"REWORK","_number":2,"created":"2020-11-25 17:05:08.000000000","uploader":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"ref":"refs/changes/89/763889/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/89/763889/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/89/763889/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/89/763889/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/89/763889/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/89/763889/2"}}},"commit":{"parents":[{"commit":"f52ca2fe5fe16ab6e3b14237c5ccaaf9a5de64b0","subject":"Add known_hosts entries for additional Gerrit keys","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f52ca2fe5fe16ab6e3b14237c5ccaaf9a5de64b0"}]}],"author":{"name":"Matt McEuen","email":"madgin@madgin.net","date":"2020-11-23 23:22:50.000000000","tz":-360},"committer":{"name":"Matt McEuen","email":"madgin@madgin.net","date":"2020-11-25 17:04:58.000000000","tz":-360},"subject":"New Project Request: airship/vino","message":"New Project Request: airship/vino\n\nThis is a request to create the airship/vino project in opendev.\n\nViNO is based on the Kubernetes ansible operator framework,\nand will have responsibility within Airship-based Kubernetes\nclusters for lifecycle managing static libvirt definitions on\nworker nodes, for use within simple CAPI sub-clusters.\n\nIf you can seed the new gerrit group with me and I can add the appropriate\ninitial core team.\n\nChange-Id: I9facf665936f9ba8529af2457bb97fc5cc6e49b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2a846c9483597f3398adf8e3909824e3480b6316"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2a846c9483597f3398adf8e3909824e3480b6316"}]},"branch":"refs/heads/master"},"62b272af4504bd75fd236b63c957f31768594f80":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-11-25 17:33:46.000000000","uploader":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"ref":"refs/changes/89/763889/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/89/763889/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/89/763889/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/89/763889/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/89/763889/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/89/763889/3"}}},"commit":{"parents":[{"commit":"f52ca2fe5fe16ab6e3b14237c5ccaaf9a5de64b0","subject":"Add known_hosts entries for additional Gerrit keys","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f52ca2fe5fe16ab6e3b14237c5ccaaf9a5de64b0"}]}],"author":{"name":"Matt McEuen","email":"madgin@madgin.net","date":"2020-11-23 23:22:50.000000000","tz":-360},"committer":{"name":"Matt McEuen","email":"madgin@madgin.net","date":"2020-11-25 17:33:28.000000000","tz":-360},"subject":"New Project Request: airship/vino","message":"New Project Request: airship/vino\n\nThis is a request to create the airship/vino project in opendev.\n\nViNo is a Kubernetes operator that has\nresponsibility within Airship-based Kubernetes\nclusters for lifecycle managing static libvirt definitions on\nworker nodes, for use within simple CAPI sub-clusters.\n\nIf you can seed the new gerrit group with me and I can add the appropriate\ninitial core team.\n\nChange-Id: I9facf665936f9ba8529af2457bb97fc5cc6e49b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/62b272af4504bd75fd236b63c957f31768594f80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/62b272af4504bd75fd236b63c957f31768594f80"}]},"branch":"refs/heads/master"},"b50da3cd7082f1aef0c81668722149205184b56a":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-12-07 17:03:12.000000000","uploader":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"ref":"refs/changes/89/763889/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/89/763889/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/89/763889/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/89/763889/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/89/763889/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/89/763889/4"}}},"commit":{"parents":[{"commit":"390944b2a9a9e58c2908be712bf9bc4235d20e12","subject":"Add \u0027Abandon\u0027 permission for retired repo ACL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/390944b2a9a9e58c2908be712bf9bc4235d20e12"}]}],"author":{"name":"Matt McEuen","email":"madgin@madgin.net","date":"2020-11-23 23:22:50.000000000","tz":-360},"committer":{"name":"Matt McEuen","email":"matt.mceuen@att.com","date":"2020-12-07 17:03:12.000000000","tz":0},"subject":"New Project Request: airship/vino","message":"New Project Request: airship/vino\n\nThis is a request to create the airship/vino project in opendev.\n\nViNo is a Kubernetes operator that has\nresponsibility within Airship-based Kubernetes\nclusters for lifecycle managing static libvirt definitions on\nworker nodes, for use within simple CAPI sub-clusters.\n\nIf you can seed the new gerrit group with me and I can add the appropriate\ninitial core team.\n\nChange-Id: I9facf665936f9ba8529af2457bb97fc5cc6e49b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b50da3cd7082f1aef0c81668722149205184b56a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b50da3cd7082f1aef0c81668722149205184b56a"}]},"branch":"refs/heads/master","description":"Rebase"},"1d3632d9361ca244999b60b4edffef012a2f2b48":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-12-08 16:35:08.000000000","uploader":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"ref":"refs/changes/89/763889/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/89/763889/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/89/763889/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/89/763889/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/89/763889/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/89/763889/5"}}},"commit":{"parents":[{"commit":"c21b2e136db31e2ce354d5d4c56c72ec443b909c","subject":"Merge \"Announce openstack/ossa in #openstack-security\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c21b2e136db31e2ce354d5d4c56c72ec443b909c"}]}],"author":{"name":"Matt McEuen","email":"madgin@madgin.net","date":"2020-11-23 23:22:50.000000000","tz":-360},"committer":{"name":"Matt McEuen","email":"matt.mceuen@att.com","date":"2020-12-08 16:35:08.000000000","tz":0},"subject":"New Project Request: airship/vino","message":"New Project Request: airship/vino\n\nThis is a request to create the airship/vino project in opendev.\n\nViNo is a Kubernetes operator that has\nresponsibility within Airship-based Kubernetes\nclusters for lifecycle managing static libvirt definitions on\nworker nodes, for use within simple CAPI sub-clusters.\n\nIf you can seed the new gerrit group with me and I can add the appropriate\ninitial core team.\n\nChange-Id: I9facf665936f9ba8529af2457bb97fc5cc6e49b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1d3632d9361ca244999b60b4edffef012a2f2b48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1d3632d9361ca244999b60b4edffef012a2f2b48"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
