)]}'
{"id":"openstack%2Fproject-config~765777","triplet_id":"openstack%2Fproject-config~master~Ia9166ad7a19fdf3fe8c3eb4b4037d5db4cbe56d0","project":"openstack/project-config","branch":"master","topic":"osa/deprecate_galera_client","hashtags":[],"change_id":"Ia9166ad7a19fdf3fe8c3eb4b4037d5db4cbe56d0","subject":"Deprecate openstack-ansible-galera_client role","status":"MERGED","created":"2020-12-07 10:28:20.000000000","updated":"2020-12-11 18:00:44.000000000","submitted":"2020-12-11 17:28:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765777-osa/deprecate_galera_client","meta_rev_id":"e719bc6b80a148242a7bb16960f1cc3a1b233832","_number":765777,"virtual_id_number":765777,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-11 17:28:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-09 15:18:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2020-12-11 17:16:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":1,"date":"2020-12-11 17:16:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-07 10:29:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-09 15:18:31.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-12-11 17:16:26.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"bb2967aaaa8ab9d116df7c088f3effc7422d138c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-07 10:28:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5229599b7edd3bdc751d2abe6b5c68f0741580db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-07 10:29:07.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  while parsing a block mapping\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 3580, column 5\ndid not find expected key\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 3587, column 7","accounts_in_message":[],"_revision_number":1},{"id":"7f71f4b1dc1484e2cd9f2905788d5bac3423d822","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-07 11:02:03.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"fb53c5c9748eea43402585aab45a3ec28ae7438f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-07 11:02:30.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  while parsing a block mapping\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 3580, column 5\ndid not find expected key\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 3587, column 7","accounts_in_message":[],"_revision_number":2},{"id":"6e416a0174201807905be646cbd84808e01ea890","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-07 11:04:46.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"da71aaab46b908fda9db48d8ee102f6c98b27c1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-07 11:05:10.000000000","message":"Patch Set 3: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  while scanning a simple key\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 22, column 5\ncould not find expected \u0027:\u0027\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 23, column 12","accounts_in_message":[],"_revision_number":3},{"id":"e1ec4aab7510d728d283a3c965f232f88062e612","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-07 11:09:15.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"29f14fe3ca97d02ebe1f23e2a47cdaa546cdcb5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-07 11:22:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a4571c95a16546059636039158d55e75 : SUCCESS in 5m 58s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/45f339dc6a5b4c38bbaadae23220e82f : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":4},{"id":"48ab44010c99ff90682cf0d2dead53270022310e","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-12-09 15:18:31.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"605bbb8eac94c3327278df64f1d210af0596d2d1","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-12-11 17:16:26.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"cf332703155cfbc0371cf2695302ba18ca5f801d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 17:16:54.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"358d98837dcd1c4831647751050d1c0a2140b5d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 17:28:28.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d83c1b88e706462d91221930708b8ca6 : SUCCESS in 6m 05s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f2f9316b814040a49d559d9d424fed7d : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":4},{"id":"bef7a68358a93639c840ce92d9ccd3e943f08853","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 17:28:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"e719bc6b80a148242a7bb16960f1cc3a1b233832","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 18:00:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/6d47e34f31454600ab108ddd0cb964fe : SUCCESS in 7m 14s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"23f8a397ac8a07f31d0d28ef4c11257579500724","revisions":{"9379f5a8fe0535d0e9d095e28e1c61f90844f26a":{"kind":"REWORK","_number":1,"created":"2020-12-07 10:28:20.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/77/765777/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/77/765777/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/765777/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/765777/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/765777/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/77/765777/1"}}},"commit":{"parents":[{"commit":"d216a051dd87f3e4ddd93cd576e7909532d30ea9","subject":"Merge \"Remove legacy-tempest-dsvm-neutron-{ipv6only,serviceipv6}\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d216a051dd87f3e4ddd93cd576e7909532d30ea9"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-12-07 10:25:10.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-12-07 10:27:57.000000000","tz":120},"subject":"Deprecate openstack-ansible-galera_client role","message":"Deprecate openstack-ansible-galera_client role\n\nChange-Id: Ia9166ad7a19fdf3fe8c3eb4b4037d5db4cbe56d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9379f5a8fe0535d0e9d095e28e1c61f90844f26a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9379f5a8fe0535d0e9d095e28e1c61f90844f26a"}]},"branch":"refs/heads/master"},"040054c35a60c4141d9ccb0fe59aac5b1ef6c418":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-12-07 11:02:03.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/77/765777/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/77/765777/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/765777/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/765777/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/765777/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/77/765777/2"}}},"commit":{"parents":[{"commit":"390944b2a9a9e58c2908be712bf9bc4235d20e12","subject":"Add \u0027Abandon\u0027 permission for retired repo ACL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/390944b2a9a9e58c2908be712bf9bc4235d20e12"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-12-07 10:25:10.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-12-07 11:02:03.000000000","tz":0},"subject":"Deprecate openstack-ansible-galera_client role","message":"Deprecate openstack-ansible-galera_client role\n\nChange-Id: Ia9166ad7a19fdf3fe8c3eb4b4037d5db4cbe56d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/040054c35a60c4141d9ccb0fe59aac5b1ef6c418"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/040054c35a60c4141d9ccb0fe59aac5b1ef6c418"}]},"branch":"refs/heads/master","description":"Rebase"},"f0cc9f4bb43754b3ec916a054d8c64429afe8fc9":{"kind":"REWORK","_number":3,"created":"2020-12-07 11:04:46.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/77/765777/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/77/765777/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/765777/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/765777/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/765777/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/77/765777/3"}}},"commit":{"parents":[{"commit":"390944b2a9a9e58c2908be712bf9bc4235d20e12","subject":"Add \u0027Abandon\u0027 permission for retired repo ACL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/390944b2a9a9e58c2908be712bf9bc4235d20e12"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-12-07 10:25:10.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-12-07 11:04:40.000000000","tz":0},"subject":"Deprecate openstack-ansible-galera_client role","message":"Deprecate openstack-ansible-galera_client role\n\nChange-Id: Ia9166ad7a19fdf3fe8c3eb4b4037d5db4cbe56d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f0cc9f4bb43754b3ec916a054d8c64429afe8fc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f0cc9f4bb43754b3ec916a054d8c64429afe8fc9"}]},"branch":"refs/heads/master"},"23f8a397ac8a07f31d0d28ef4c11257579500724":{"kind":"REWORK","_number":4,"created":"2020-12-07 11:09:15.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/77/765777/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/77/765777/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/765777/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/765777/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/765777/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/77/765777/4"}}},"commit":{"parents":[{"commit":"390944b2a9a9e58c2908be712bf9bc4235d20e12","subject":"Add \u0027Abandon\u0027 permission for retired repo ACL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/390944b2a9a9e58c2908be712bf9bc4235d20e12"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-12-07 10:25:10.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-12-07 11:09:10.000000000","tz":0},"subject":"Deprecate openstack-ansible-galera_client role","message":"Deprecate openstack-ansible-galera_client role\n\nChange-Id: Ia9166ad7a19fdf3fe8c3eb4b4037d5db4cbe56d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/23f8a397ac8a07f31d0d28ef4c11257579500724"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/23f8a397ac8a07f31d0d28ef4c11257579500724"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
