)]}'
{"id":"openstack%2Fproject-config~767057","triplet_id":"openstack%2Fproject-config~master~I77a3daa1a9aa6146d6ff3c4f2f1dd4716e01ff56","project":"openstack/project-config","branch":"master","topic":"retire-karbor","hashtags":[],"change_id":"I77a3daa1a9aa6146d6ff3c4f2f1dd4716e01ff56","subject":"Remove Karbor projects from infra","status":"MERGED","created":"2020-12-15 00:30:43.000000000","updated":"2021-01-29 07:27:05.000000000","submitted":"2021-01-29 06:41:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"767057-retire-karbor","meta_rev_id":"c4254d46ccd199b0f8360e31472affeaa52c27c7","_number":767057,"virtual_id_number":767057,"owner":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-29 06:41:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2021-01-28 19:34:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2021-01-28 20:01:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-29 06:28:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-29 06:28:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-15 00:31:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-21 07:48:58.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2021-01-28 18:31:59.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2021-01-28 20:01:42.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-01-29 06:28:42.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"}],"messages":[{"id":"027ce7b6f99f2f51ff9243ed3c52f54ed8391e0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-12-15 00:30:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd4da3aaa74c461ff42d3df8cb1a589dd1dec6d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 00:31:03.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":1},{"id":"734bea5a582a94125724c9adb29bc9881702d83d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-12-15 00:31:04.000000000","message":"Topic retire-karbor2 removed","accounts_in_message":[],"_revision_number":1},{"id":"c4aa75fdce65bf6c15360a4a2e02156a366778b3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-12-15 00:31:12.000000000","message":"Topic set to retire-karbor","accounts_in_message":[],"_revision_number":1},{"id":"f3352cb5b96c75b48bea758aa851e9c48c74994d","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-12-15 23:42:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1ba82e9076b567abc3f687745c31dea4624da855","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 23:42:49.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"5004dd1396e4077de65083b22f608b85b378454e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2021-01-20 20:12:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ce38a7d4ab07b4f40d3ca1819ad2a05ddf42e39f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2021-01-20 20:12:40.000000000","message":"Topic retire-karbor2 removed","accounts_in_message":[],"_revision_number":2},{"id":"550f2995b405ad39d7174b22405aa21bdd4b9c56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 20:12:42.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  while scanning a simple key\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 4947, column 1\ncould not find expected \u0027:\u0027\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 4948, column 9","accounts_in_message":[],"_revision_number":2},{"id":"2a5aefdd6ce1df974705aa157045ff94df17388e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2021-01-20 20:13:12.000000000","message":"Topic set to retire-karbor","accounts_in_message":[],"_revision_number":2},{"id":"bfea34230b816000f86697ee7172ba67aedad7d1","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-01-21 07:48:58.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"394626844afce59079b1f690915ec033c813feda","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-28 18:31:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3feec24bc8ab89856f25b351cdfb0d2b6ed75fb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-28 18:34:04.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"03982940ca43627342a658e42271d13cc4adf39f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 18:53:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/cc8ba6274c71448da902ffd0d2d0e104 : SUCCESS in 6m 25s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/3611e80137c842c5991ba6d8270da72a : SUCCESS in 3m 55s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/c9cfab68b1c64f0699f6d26a73ffff74 : SUCCESS in 6m 47s","accounts_in_message":[],"_revision_number":3},{"id":"2ab42f270036d0d9d382864d356a229c0a23e8bc","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-01-28 19:34:42.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"60662fc385c185c287ae814d143180d0ac6d0652","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-28 20:01:42.000000000","message":"Patch Set 3: Code-Review+1\n\nlgtm, I just fixed rebase conflict.","accounts_in_message":[],"_revision_number":3},{"id":"70887127d16c23615e49053bea6928b16cff05ad","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-01-29 06:28:42.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9a8306578aac3e39e8df200f35d0e2b5815ad1bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 06:29:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0e337651860dd309dd28349ec74681254d74ad03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 06:41:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a61614e6960b46a5bf3c4c494398fe23 : SUCCESS in 4m 49s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/2e1c1d069cec40a6bdd3eb687663fb0a : SUCCESS in 3m 47s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/78e866e351ca4e4393e4c1f7e71231b5 : SUCCESS in 7m 19s","accounts_in_message":[],"_revision_number":3},{"id":"20c2610e1bb6f10e4555e234e3e3a7ae7027e77f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 06:41:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"c4254d46ccd199b0f8360e31472affeaa52c27c7","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 07:27:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/75d18da3d6044cde961107ca85e462d1 : SUCCESS in 10m 28s\n- infra-prod-service-eavesdrop https://zuul.opendev.org/t/openstack/build/a2e3aef04c3c407e9a59fb31e1de5078 : SUCCESS in 7m 31s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"67b5a9d2ee7331f6f0325e994b8f5152574c9b30","revisions":{"73401929fea3024885f29a1660787659f43be20d":{"kind":"REWORK","_number":1,"created":"2020-12-15 00:30:43.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/57/767057/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/767057/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/767057/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/767057/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/767057/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/767057/1"}}},"commit":{"parents":[{"commit":"63cdece7f0bbca55b3f5a36ca164ee4bf7fe2807","subject":"Merge \"Re-add os_monasca roles to zuul\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/63cdece7f0bbca55b3f5a36ca164ee4bf7fe2807"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-12-15 00:27:55.000000000","tz":-480},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-12-15 00:27:55.000000000","tz":-480},"subject":"Remove Karbor projects from infra","message":"Remove Karbor projects from infra\n\nAs announced on the openstack-discuss ML[1], Karbor is retiring\nthis cycle (Wallaby).\n\nThis commit retires this repository as per the process defined in\nthe project-guide[2].\n\nThank you to all the contributors of Karbor for your hard work!\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-November/018643.html\n[2] https://docs.openstack.org/project-team-guide/repository.html#retiring-a-repository\n\nDepends-On: https://review.opendev.org/c/openstack/karbor/+/767032\nDepends-On: https://review.opendev.org/c/openstack/karbor-dashboard/+/767034\nDepends-On: https://review.opendev.org/c/openstack/python-karborclient/+/767033\nChange-Id: I77a3daa1a9aa6146d6ff3c4f2f1dd4716e01ff56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/73401929fea3024885f29a1660787659f43be20d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/73401929fea3024885f29a1660787659f43be20d"}]},"branch":"refs/heads/master"},"f0d849e53b3e6db29d8ab21b63a1b453fe46d15a":{"kind":"REWORK","_number":2,"created":"2021-01-20 20:12:19.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/57/767057/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/767057/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/767057/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/767057/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/767057/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/767057/2"}}},"commit":{"parents":[{"commit":"104a26e69fe7dfc826ddaf908047b4825210493d","subject":"Merge \"Move snaps ACL to x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/104a26e69fe7dfc826ddaf908047b4825210493d"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-12-15 00:27:55.000000000","tz":-480},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-01-20 20:12:08.000000000","tz":-480},"subject":"Remove Karbor projects from infra","message":"Remove Karbor projects from infra\n\nAs announced on the openstack-discuss ML[1], Karbor is retiring\nthis cycle (Wallaby).\n\nThis commit retires this repository as per the process defined in\nthe project-guide[2].\n\nThank you to all the contributors of Karbor for your hard work!\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-November/018643.html\n[2] https://docs.openstack.org/project-team-guide/repository.html#retiring-a-repository\n\nDepends-On: https://review.opendev.org/c/openstack/karbor/+/767032\nDepends-On: https://review.opendev.org/c/openstack/karbor-dashboard/+/767034\nDepends-On: https://review.opendev.org/c/openstack/python-karborclient/+/767033\nChange-Id: I77a3daa1a9aa6146d6ff3c4f2f1dd4716e01ff56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f0d849e53b3e6db29d8ab21b63a1b453fe46d15a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f0d849e53b3e6db29d8ab21b63a1b453fe46d15a"}]},"branch":"refs/heads/master"},"67b5a9d2ee7331f6f0325e994b8f5152574c9b30":{"kind":"REWORK","_number":3,"created":"2021-01-28 18:34:04.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/57/767057/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/767057/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/767057/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/767057/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/767057/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/767057/3"}}},"commit":{"parents":[{"commit":"104a26e69fe7dfc826ddaf908047b4825210493d","subject":"Merge \"Move snaps ACL to x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/104a26e69fe7dfc826ddaf908047b4825210493d"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-12-15 00:27:55.000000000","tz":-480},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2021-01-28 18:33:32.000000000","tz":0},"subject":"Remove Karbor projects from infra","message":"Remove Karbor projects from infra\n\nAs announced on the openstack-discuss ML[1], Karbor is retiring\nthis cycle (Wallaby).\n\nThis commit retires this repository as per the process defined in\nthe project-guide[2].\n\nThank you to all the contributors of Karbor for your hard work!\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-November/018643.html\n[2] https://docs.openstack.org/project-team-guide/repository.html#retiring-a-repository\n\nDepends-On: https://review.opendev.org/c/openstack/karbor/+/767032\nDepends-On: https://review.opendev.org/c/openstack/karbor-dashboard/+/767034\nDepends-On: https://review.opendev.org/c/openstack/python-karborclient/+/767033\nChange-Id: I77a3daa1a9aa6146d6ff3c4f2f1dd4716e01ff56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/67b5a9d2ee7331f6f0325e994b8f5152574c9b30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/67b5a9d2ee7331f6f0325e994b8f5152574c9b30"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
