)]}'
{"id":"openstack%2Fproject-config~767062","triplet_id":"openstack%2Fproject-config~master~I208b4e7dcefb1b697c2a0ff25d2c051d4b8593ed","project":"openstack/project-config","branch":"master","topic":"retire-karbor","hashtags":[],"change_id":"I208b4e7dcefb1b697c2a0ff25d2c051d4b8593ed","subject":"Update retiring Karbor repo ACL to openstack/retired.config","status":"MERGED","created":"2020-12-15 01:15:45.000000000","updated":"2020-12-16 03:18:46.000000000","submitted":"2020-12-16 03:03:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"767062-retire-karbor","meta_rev_id":"94253bd73d93d8d5e679f13bdf33bb8a65d6f8aa","_number":767062,"virtual_id_number":767062,"owner":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-16 03:03:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-12-15 15:25:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-16 02:49:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-16 02:49:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-15 01:23:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-15 15:25:07.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-12-16 02:49:50.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"}],"messages":[{"id":"873fe0b5320fe7d7df73c2276f34a7f8a2420e0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-12-15 01:15:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c1acfbb012452bd2f00c5e07344c6113b3667c5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-12-15 01:16:00.000000000","message":"Topic retire-karbor3 removed","accounts_in_message":[],"_revision_number":1},{"id":"7aaa1a4f1ae8affc3050c2972b63bf7293aab20a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-12-15 01:16:05.000000000","message":"Topic set to retire-karbor","accounts_in_message":[],"_revision_number":1},{"id":"a778684de4ed5582e3f193d2bb2aef4bafab9f12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 01:23:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ac2ba80c8b6c4da6b76139f3800222d4 : SUCCESS in 6m 12s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/6bef18bbdd5b47b0b8ad27d87a490d06 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":1},{"id":"39b8f0b83f2c2b7f5e8d577f289ed71b081afa89","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-12-15 15:25:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fdb6875941f8299418cbd41fdeea337adec0f256","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-12-16 02:49:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b5285d50868b65e0ed10da9966870c977d64644b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 02:50:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"47cad4bfefdb93b2e7e51aeadd9b61d0ab1e5881","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 03:03:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1b3a300d385145c8a115b07e004a5623 : SUCCESS in 5m 06s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/5953f5a3f7ba4434a8f067402a0fde00 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"8c89d570b8e9d68c96e62771a8bdd208ad155e51","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 03:03:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"94253bd73d93d8d5e679f13bdf33bb8a65d6f8aa","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 03:18:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/f6f9ce2af1d74eb792bab30cb02ba484 : SUCCESS in 7m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ec5eee334066fc71094d635bec675f4f629aaaa6","revisions":{"ec5eee334066fc71094d635bec675f4f629aaaa6":{"kind":"REWORK","_number":1,"created":"2020-12-15 01:15:45.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/62/767062/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/62/767062/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/767062/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/767062/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/767062/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/62/767062/1"}}},"commit":{"parents":[{"commit":"63cdece7f0bbca55b3f5a36ca164ee4bf7fe2807","subject":"Merge \"Re-add os_monasca roles to zuul\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/63cdece7f0bbca55b3f5a36ca164ee4bf7fe2807"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-12-15 01:05:10.000000000","tz":-480},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-12-15 01:14:25.000000000","tz":-480},"subject":"Update retiring Karbor repo ACL to openstack/retired.config","message":"Update retiring Karbor repo ACL to openstack/retired.config\n\nAs announced on the openstack-discuss ML[1], Karbor is retiring\nthis cycle (Wallaby).\n\nThis commit updated the ACL for Karbor repos to\nopenstack/retired.config so that the TC can finish the removal\nof the code and close open reviews.\n\nNeeded-By: https://review.opendev.org/c/openstack/karbor/+/767032\nNeeded-By: https://review.opendev.org/c/openstack/karbor-dashboard/+/767034\nNeeded-By: https://review.opendev.org/c/openstack/python-karborclient/+/767033\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-November/018643.html\n\nChange-Id: I208b4e7dcefb1b697c2a0ff25d2c051d4b8593ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ec5eee334066fc71094d635bec675f4f629aaaa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ec5eee334066fc71094d635bec675f4f629aaaa6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
