)]}'
{"id":"openstack%2Fproject-config~771073","triplet_id":"openstack%2Fproject-config~master~If407e4c4fa9d81c13577bac6240e02472a7fb6b2","project":"openstack/project-config","branch":"master","topic":"openinfra-channel","hashtags":[],"change_id":"If407e4c4fa9d81c13577bac6240e02472a7fb6b2","subject":"Set up access for #openinfra channel","status":"MERGED","created":"2021-01-15 19:55:41.000000000","updated":"2021-01-22 02:07:36.000000000","submitted":"2021-01-22 01:57:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"771073-openinfra-channel","meta_rev_id":"ce28653f2b95bf4374a051f17ba39000ec686bd3","_number":771073,"virtual_id_number":771073,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-22 01:57:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2021-01-18 07:31:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-22 01:49:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-15 20:13:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-18 07:31:55.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"61b3a7bf2c12125bef3434fc7bf5e7d2d6be0d63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-01-15 19:55:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"824abe295481922a173f615a594123675925746d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 20:13:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b54db869b82e4aaf8cc0de7142e82bdf : SUCCESS in 6m 13s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/af73b21712c641408acad2ae75190c2e : FAILURE in 7m 29s","accounts_in_message":[],"_revision_number":1},{"id":"e6d880d136c8445a65dbcc3f63d7a68c8b39bbfa","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-01-15 20:53:03.000000000","message":"Patch Set 1: Workflow-1\n\nLooks like the permissions Lyz had to bestow on us were insufficient, and we\u0027ll need to get Nibz to add FRi to our flags.","accounts_in_message":[],"_revision_number":1},{"id":"f528857f6a57f1371578b518ecda9a9b7600ad25","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-01-18 07:31:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1604ca01b270d70204be69cac7d2d4205028162b","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-01-22 01:32:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ced818d40030897f06a7f35289b0d15b8ffbf847","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 01:42:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f81780b2adf74563afa819103f6ca4b6 : SUCCESS in 6m 37s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/5e941756b1b946019915483dfd6f1c75 : SUCCESS in 7m 57s","accounts_in_message":[],"_revision_number":1},{"id":"2cc68560ad61f388b27e54153c6dfa36406fa6c2","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-01-22 01:49:08.000000000","message":"Patch Set 1: Workflow+1\n\nYay! We\u0027re in.","accounts_in_message":[],"_revision_number":1},{"id":"e680ea13bbbf3fd93aa920cc0cb30c6a5da6ecdc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 01:49:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1fab181f3619485dbdb83aba0a72e4b817a02976","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 01:57:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4521bd5640424bcc875ec85f9be44a77 : SUCCESS in 6m 03s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/809d25511aee43239fe7697235ee7e6e : SUCCESS in 7m 14s","accounts_in_message":[],"_revision_number":1},{"id":"61d13ade9a7d78f4b44391ef8d0637eeda87dc0b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 01:57:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ce28653f2b95bf4374a051f17ba39000ec686bd3","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 02:07:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-run-accessbot https://zuul.opendev.org/t/openstack/build/828d248a5ce146429bd2582804f47931 : SUCCESS in 9m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"82a2e6a629f06896610f7f6dffd59cc0e0e6edba","revisions":{"82a2e6a629f06896610f7f6dffd59cc0e0e6edba":{"kind":"REWORK","_number":1,"created":"2021-01-15 19:55:41.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/73/771073/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/73/771073/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/771073/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/771073/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/771073/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/73/771073/1"}}},"commit":{"parents":[{"commit":"083ddaf2f97bda2144ff867ea613f69741efd3b4","subject":"Un-pause Gentoo image builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/083ddaf2f97bda2144ff867ea613f69741efd3b4"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-01-15 19:53:15.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-01-15 19:53:15.000000000","tz":0},"subject":"Set up access for #openinfra channel","message":"Set up access for #openinfra channel\n\nThe Open Infrastructure Foundation is going to work on consolidating\nits IRC channel names under a common namespace, #openinfra on\nFreenode. Access has been granted to our bot account by the original\nchannel operators.\n\nChange-Id: If407e4c4fa9d81c13577bac6240e02472a7fb6b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/82a2e6a629f06896610f7f6dffd59cc0e0e6edba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/82a2e6a629f06896610f7f6dffd59cc0e0e6edba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
