)]}'
{"id":"openstack%2Fproject-config~771235","triplet_id":"openstack%2Fproject-config~master~I7e9944f3f9275a1b48e421c938b13aad48dc01f7","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I7e9944f3f9275a1b48e421c938b13aad48dc01f7","subject":"Add PTP Notification app to StarlingX","status":"MERGED","created":"2021-01-18 15:33:52.000000000","updated":"2021-01-19 19:14:18.000000000","submitted":"2021-01-19 15:22:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"771235","meta_rev_id":"165be9587608ab7b75984a46b92dc475dfc1b3e1","_number":771235,"virtual_id_number":771235,"owner":{"_account_id":31928,"name":"Cole Walker","email":"cole.walker@windriver.com","username":"cwalker"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26026,"name":"Greg Waines","email":"greg.waines@windriver.com","username":"gwaines"},{"value":0,"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-19 15:22:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"all":[{"value":1,"date":"2021-01-18 17:21:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26026,"name":"Greg Waines","email":"greg.waines@windriver.com","username":"gwaines"},{"value":1,"date":"2021-01-19 14:34:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-19 15:10:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":26026,"name":"Greg Waines","email":"greg.waines@windriver.com","username":"gwaines"},{"value":0,"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-19 15:10:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"reviewers":{"CC":[{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"}],"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26026,"name":"Greg Waines","email":"greg.waines@windriver.com","username":"gwaines"},{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-18 15:39:59.000000000","updated_by":{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},"reviewer":{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},"state":"CC"},{"updated":"2021-01-18 15:39:59.000000000","updated_by":{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},"reviewer":{"_account_id":26026,"name":"Greg Waines","email":"greg.waines@windriver.com","username":"gwaines"},"state":"REVIEWER"},{"updated":"2021-01-18 15:48:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-18 17:22:56.000000000","updated_by":{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2021-01-18 17:22:56.000000000","updated_by":{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2021-01-19 04:45:50.000000000","updated_by":{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2021-01-19 14:34:44.000000000","updated_by":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"reviewer":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"state":"REVIEWER"},{"updated":"2021-01-19 15:10:08.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"8a003cd9aafb6db105a96da1f62774a20baf229f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31928,"name":"Cole Walker","email":"cole.walker@windriver.com","username":"cwalker"},"date":"2021-01-18 15:33:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a92cb79ed378dbfd01b4e3bec85775a9a584999a","author":{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},"date":"2021-01-18 15:39:59.000000000","message":"Patch Set 1:\n\nAdding Greg Waines, StarlingX TSC, so that he can +1 so that the opendev infra team can proceed with the request.","accounts_in_message":[],"_revision_number":1},{"id":"5790d92190dbb8cac968c1035fb893bfdeb44a95","author":{"_account_id":26026,"name":"Greg Waines","email":"greg.waines@windriver.com","username":"gwaines"},"date":"2021-01-18 15:42:29.000000000","message":"Patch Set 1: Code-Review+1\n\nlgtm","accounts_in_message":[],"_revision_number":1},{"id":"b689068d7877ab0504ecab98d8b54406fdc2f710","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 15:48:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/49b08fde7ec24b1e9b8aa271a881171c : SUCCESS in 5m 47s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/5f4e75d7e7a34a2283bec89cd89b4e75 : FAILURE in 3m 30s","accounts_in_message":[],"_revision_number":1},{"id":"b23ebac9f508b44fc3db666f14f15b6d0684cc94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31928,"name":"Cole Walker","email":"cole.walker@windriver.com","username":"cwalker"},"date":"2021-01-18 15:55:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"28879fc5743b0edb2cc6ede3f262872e1186645c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 16:12:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ca691ccfca664f5f9811dd1713669d65 : SUCCESS in 6m 55s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/bcf84fd134824e07a27b206b9dca6213 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":2},{"id":"dd1068f5cf147db5d76889790735294866f9864d","author":{"_account_id":26026,"name":"Greg Waines","email":"greg.waines@windriver.com","username":"gwaines"},"date":"2021-01-18 17:21:49.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2a25dfc8c2a63bf9a2f7c6b3b2906b4fc0281299","author":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"date":"2021-01-19 14:34:44.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"51432226ba86da3228c6cb819eb35cba416dc606","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-01-19 15:10:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e3cba0422cdfb33ea00cdd64c25df5012fff7591","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 15:10:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2a9f74305e58ebec0d519bb6cf38b8a09075d85b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 15:22:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/bf644cf2063540598fdc4ae9ff6653f9 : SUCCESS in 6m 30s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/cb8b24b0f0c44fe5a370eeb9bb86b285 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":2},{"id":"fdd69c70cbb6aa6cf89e9ee44bedc61bf298e408","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 15:22:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2aa8ee5390ab16961229b0b4bdb115493e25053c","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 15:27:36.000000000","message":"Patch Set 2:\n\nBuild failed (deploy pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/ea9b8fbb985b456fa0721f71de6f4f44 : RETRY_LIMIT in 55s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":2},{"id":"5e7aa3a6c7171dec5c28682093140e905c0cac95","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 18:13:56.000000000","message":"Patch Set 2:\n\nBuild failed (deploy pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/5e954c5436c241979305560cc8826f9d : RETRY_LIMIT in 1m 32s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":2},{"id":"165be9587608ab7b75984a46b92dc475dfc1b3e1","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 19:14:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/52d1d0ae54f14a7aac208fc906a79669 : SUCCESS in 7m 53s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/6263594f7c954327ada2d5ce26f0e668 : SUCCESS in 16m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0b0281c5525bd0f65c0f2543f8c6c816bc25336f","revisions":{"b82f3c01237bcb1230ab224a15c60b6357a960e6":{"kind":"REWORK","_number":1,"created":"2021-01-18 15:33:52.000000000","uploader":{"_account_id":31928,"name":"Cole Walker","email":"cole.walker@windriver.com","username":"cwalker"},"ref":"refs/changes/35/771235/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/35/771235/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/771235/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/771235/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/771235/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/35/771235/1"}}},"commit":{"parents":[{"commit":"fb00c509f86f42e19215edff913211cfeee38994","subject":"Merge \"add gnupg2 to package map for gentoo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fb00c509f86f42e19215edff913211cfeee38994"}]}],"author":{"name":"Cole Walker","email":"cole.walker@windriver.com","date":"2021-01-18 15:29:55.000000000","tz":-300},"committer":{"name":"Cole Walker","email":"cole.walker@windriver.com","date":"2021-01-18 15:33:33.000000000","tz":-300},"subject":"Add PTP Notification app to StarlingX","message":"Add PTP Notification app to StarlingX\n\nSigned-off-by: Cole Walker \u003ccole.walker@windriver.com\u003e\nChange-Id: I7e9944f3f9275a1b48e421c938b13aad48dc01f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b82f3c01237bcb1230ab224a15c60b6357a960e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b82f3c01237bcb1230ab224a15c60b6357a960e6"}]},"branch":"refs/heads/master"},"0b0281c5525bd0f65c0f2543f8c6c816bc25336f":{"kind":"REWORK","_number":2,"created":"2021-01-18 15:55:58.000000000","uploader":{"_account_id":31928,"name":"Cole Walker","email":"cole.walker@windriver.com","username":"cwalker"},"ref":"refs/changes/35/771235/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/35/771235/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/771235/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/771235/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/771235/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/35/771235/2"}}},"commit":{"parents":[{"commit":"fb00c509f86f42e19215edff913211cfeee38994","subject":"Merge \"add gnupg2 to package map for gentoo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fb00c509f86f42e19215edff913211cfeee38994"}]}],"author":{"name":"Cole Walker","email":"cole.walker@windriver.com","date":"2021-01-18 15:29:55.000000000","tz":-300},"committer":{"name":"Cole Walker","email":"cole.walker@windriver.com","date":"2021-01-18 15:52:48.000000000","tz":-300},"subject":"Add PTP Notification app to StarlingX","message":"Add PTP Notification app to StarlingX\n\nSigned-off-by: Cole Walker \u003ccole.walker@windriver.com\u003e\nChange-Id: I7e9944f3f9275a1b48e421c938b13aad48dc01f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0b0281c5525bd0f65c0f2543f8c6c816bc25336f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0b0281c5525bd0f65c0f2543f8c6c816bc25336f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
