)]}'
{"id":"openstack%2Fproject-config~771565","triplet_id":"openstack%2Fproject-config~master~Ibb64ff8dbdac9eb483dc9bb9397bd3bed9a04293","project":"openstack/project-config","branch":"master","topic":"story/2007196","hashtags":[],"change_id":"Ibb64ff8dbdac9eb483dc9bb9397bd3bed9a04293","subject":"Add ubuntu bionic and focal with xxxlarge flavor","status":"MERGED","created":"2021-01-20 07:30:17.000000000","updated":"2021-01-21 14:55:57.000000000","submitted":"2021-01-21 14:42:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"771565-story/2007196","meta_rev_id":"8629749dc79968716b5f151e60f596ea4dfafd7c","_number":771565,"virtual_id_number":771565,"owner":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-21 14:42:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"all":[{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":1,"date":"2021-01-21 13:11:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-21 14:30:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-21 14:30:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-20 07:33:47.000000000","updated_by":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"reviewer":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"state":"REVIEWER"},{"updated":"2021-01-20 07:37:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-20 11:02:08.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2021-01-21 14:30:29.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"54c336548e6d91e2bf69f3bcdea042e6c3c57844","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2021-01-20 07:30:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d40fc5e14589e7683fe0e543140accea3d6593ba","author":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"date":"2021-01-20 07:33:47.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8649da604585b50a9098ca393eeec9d49ef35a5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 07:37:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5fdd7e9baa3242808c642a562cdc924c : SUCCESS in 5m 20s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/f711f6cb77bf411fbd060dbcbd5208b8 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"fc7337f4aba8e63c1d76df40375e789a0d5e5829","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2021-01-20 07:38:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0157cacfcbc16ec7f1a030596703259af8a33123","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-01-20 11:02:08.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nSide note: It looks weird when you review your own patch, we would generally assume that you do that before submitting it.\n\nPlease also see my remarks below.","accounts_in_message":[],"_revision_number":1},{"id":"f8a835d74d36b26486bd8450139e6de824edd187","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2021-01-21 07:01:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f3e0266c0af3dbc826f1d85eca0162b2f328ea47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 07:14:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/08c62180a829499aabb0bca81cff4265 : SUCCESS in 5m 53s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/748ba2cd22104113a4d9fbcc005270f8 : FAILURE in 4m 50s","accounts_in_message":[],"_revision_number":2},{"id":"d177cde29f0fbfe1dd2331419e2b7a5baefb5d32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2021-01-21 12:55:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"df451370f1617cdd155b58961c759c294bf5954b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 13:07:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/020c9f08ba3245fcab73279b35d53e82 : SUCCESS in 5m 20s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/ffa046a5404e4c54abc9a1abada32911 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":3},{"id":"c7a18369f5f832c48279436aa791a71a65769da8","author":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"date":"2021-01-21 13:11:14.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"0ae6abc137f77a88a7ff4e8d52f03d1311498fd7","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-01-21 14:30:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2d1407429bef582db721d973d52eea3f458f3581","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 14:30:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"dcb52be2d8e38ea733d8f067774b6cfbbab72a8e","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-01-21 14:37:44.000000000","message":"Patch Set 3:\n\nRemember we currently have only one provider for this architecture, with a fairly limited quota.","accounts_in_message":[],"_revision_number":3},{"id":"a39e721cd9940411ca4620c536edae99731e1f61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 14:42:16.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/effc63ac4fc947b4816fc937a5360fe7 : SUCCESS in 5m 27s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/8b4a2decb0f1444f815325eb0053b7ca : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":3},{"id":"ab07397d32e49fe79c4935575878d55a50f5f0bc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 14:42:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"8629749dc79968716b5f151e60f596ea4dfafd7c","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 14:55:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/528fe840315548bd948967e27b58c8b6 : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"031c748dded22f3e8389f986044106910233fddc","revisions":{"a31aa5bd91de7348b24987ae00cbccbe99377ebc":{"kind":"REWORK","_number":1,"created":"2021-01-20 07:30:17.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/65/771565/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/65/771565/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/65/771565/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/65/771565/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/65/771565/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/65/771565/1"}}},"commit":{"parents":[{"commit":"104a26e69fe7dfc826ddaf908047b4825210493d","subject":"Merge \"Move snaps ACL to x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/104a26e69fe7dfc826ddaf908047b4825210493d"}]}],"author":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2021-01-20 07:29:16.000000000","tz":480},"committer":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2021-01-20 07:29:16.000000000","tz":480},"subject":"Add ubuntu-bionic-arm64-xlarge","message":"Add ubuntu-bionic-arm64-xlarge\n\nubuntu-bionic-arm64-xlarge will use m1.xxxlarge as flavor\n\nChange-Id: Ibb64ff8dbdac9eb483dc9bb9397bd3bed9a04293\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a31aa5bd91de7348b24987ae00cbccbe99377ebc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a31aa5bd91de7348b24987ae00cbccbe99377ebc"}]},"branch":"refs/heads/master"},"5b441830cff553a57e0c3613203b44fdaafed9f8":{"kind":"REWORK","_number":2,"created":"2021-01-21 07:01:48.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/65/771565/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/65/771565/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/65/771565/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/65/771565/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/65/771565/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/65/771565/2"}}},"commit":{"parents":[{"commit":"104a26e69fe7dfc826ddaf908047b4825210493d","subject":"Merge \"Move snaps ACL to x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/104a26e69fe7dfc826ddaf908047b4825210493d"}]}],"author":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2021-01-20 07:29:16.000000000","tz":480},"committer":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2021-01-21 07:00:05.000000000","tz":480},"subject":"Add ubuntu bionic and focal with xxxlarge flavor","message":"Add ubuntu bionic and focal with xxxlarge flavor\n\nubuntu-focal-arm64-xxxlarge and ubuntu-bionic-arm64-xxxlarge\nwill use m1.xxxlarge as flavor\n\nChange-Id: Ibb64ff8dbdac9eb483dc9bb9397bd3bed9a04293\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5b441830cff553a57e0c3613203b44fdaafed9f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5b441830cff553a57e0c3613203b44fdaafed9f8"}]},"branch":"refs/heads/master"},"031c748dded22f3e8389f986044106910233fddc":{"kind":"REWORK","_number":3,"created":"2021-01-21 12:55:57.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/65/771565/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/65/771565/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/65/771565/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/65/771565/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/65/771565/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/65/771565/3"}}},"commit":{"parents":[{"commit":"104a26e69fe7dfc826ddaf908047b4825210493d","subject":"Merge \"Move snaps ACL to x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/104a26e69fe7dfc826ddaf908047b4825210493d"}]}],"author":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2021-01-20 07:29:16.000000000","tz":480},"committer":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2021-01-21 12:55:47.000000000","tz":480},"subject":"Add ubuntu bionic and focal with xxxlarge flavor","message":"Add ubuntu bionic and focal with xxxlarge flavor\n\nubuntu-focal-arm64-xxxlarge and ubuntu-bionic-arm64-xxxlarge\nwill use m1.xxxlarge as flavor\n\nChange-Id: Ibb64ff8dbdac9eb483dc9bb9397bd3bed9a04293\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/031c748dded22f3e8389f986044106910233fddc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/031c748dded22f3e8389f986044106910233fddc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
