)]}'
{"id":"openstack%2Fproject-config~771912","triplet_id":"openstack%2Fproject-config~master~Id4ba3e16d9c3865dc900f9b5e70dc705be64a6e7","project":"openstack/project-config","branch":"master","topic":"story/2007196","hashtags":[],"change_id":"Id4ba3e16d9c3865dc900f9b5e70dc705be64a6e7","subject":"Mark min-ready for ubuntu-focal-arm64","status":"ABANDONED","created":"2021-01-22 06:30:38.000000000","updated":"2021-03-18 08:18:31.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8f9500eed2240abf53eb49fa439858cd4beb80cf","_number":771912,"virtual_id_number":771912,"owner":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-01-22 06:42:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2021-01-22 07:04:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-22 06:32:18.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-22 06:32:18.000000000","updated_by":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"reviewer":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"state":"REVIEWER"},{"updated":"2021-01-22 06:42:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-22 07:04:58.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"4eff0030db07c406b66988e4676d574ed5310b74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2021-01-22 06:30:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6bf21c5545a480cacb9cbf5215a2c35dfb6303f8","author":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"date":"2021-01-22 06:32:18.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b1a3eda5e2e23f178e68710b081d4f17b98e933f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 06:42:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/482ef39a36a948d5931b9fd085f37101 : SUCCESS in 7m 29s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/cd27b0184dab42d888982d8cb0ee8409 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"c9d7a14d16ff4320730a5745023bf8442164921f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-01-22 07:04:58.000000000","message":"Patch Set 1: Code-Review-1\n\nI don\u0027t think that with our limited quota it is useful to have so many nodes pre-deployed. Can you explain in the commit message why you think this is necessary? And maybe consider lowering the min-ready for other node types at the same time?","accounts_in_message":[],"_revision_number":1},{"id":"8f9500eed2240abf53eb49fa439858cd4beb80cf","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2021-03-18 08:18:31.000000000","message":"Abandoned\n\nLet\u0027s do https://review.opendev.org/c/openstack/project-config/+/781219 and rethink about introduce this once we actually got more requirement for this label","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5ae9189b5cd1feeff5761e12ec179f7484409324","revisions":{"5ae9189b5cd1feeff5761e12ec179f7484409324":{"kind":"REWORK","_number":1,"created":"2021-01-22 06:30:38.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/12/771912/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/12/771912/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/771912/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/771912/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/771912/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/12/771912/1"}}},"commit":{"parents":[{"commit":"27ffaf3b44b28928348993b8763605125eb0aee8","subject":"Merge \"Set up access for #openinfra channel\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/27ffaf3b44b28928348993b8763605125eb0aee8"}]}],"author":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2021-01-22 06:30:08.000000000","tz":480},"committer":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2021-01-22 06:30:08.000000000","tz":480},"subject":"Mark min-ready for ubuntu-focal-arm64","message":"Mark min-ready for ubuntu-focal-arm64\n\nChange-Id: Id4ba3e16d9c3865dc900f9b5e70dc705be64a6e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5ae9189b5cd1feeff5761e12ec179f7484409324"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5ae9189b5cd1feeff5761e12ec179f7484409324"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
