)]}'
{"id":"openstack%2Fproject-config~772364","triplet_id":"openstack%2Fproject-config~master~Ic9e0f2071ac84deca60223471a35c087dde852a2","project":"openstack/project-config","branch":"master","topic":"Open10K8S","hashtags":[],"change_id":"Ic9e0f2071ac84deca60223471a35c087dde852a2","subject":"Add zuul-storage-proxy in zuul namespace","status":"MERGED","created":"2021-01-25 15:42:21.000000000","updated":"2021-02-10 17:37:01.000000000","submitted":"2021-02-10 16:40:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"772364-Open10K8S","meta_rev_id":"e1fbbc5d425fb90b18bc966f61e8dd391f80890f","_number":772364,"virtual_id_number":772364,"owner":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-10 16:40:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2021-02-10 16:26:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-10 16:18:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":1,"date":"2021-02-10 16:26:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-10 16:18:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-25 15:44:09.000000000","updated_by":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2021-01-25 15:47:19.000000000","updated_by":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2021-01-25 15:55:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-25 16:26:41.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2021-02-09 20:25:39.000000000","updated_by":{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},"reviewer":{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},"state":"REVIEWER"},{"updated":"2021-02-10 15:27:07.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"CC"},{"updated":"2021-02-10 15:56:36.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"5de6f4622cda2ff6c234d80b00478cb587c9b326","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2021-01-25 15:42:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d56174b76947de674958157935ca3a2949689a5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 15:55:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f7fd028df3e64b28a286c51f9bd132bf : SUCCESS in 5m 49s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b4a71df869ee4594a8fd985000a90963 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":1},{"id":"8d4f1c8192c9356e5e06aff9cf912eaeaa49de44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2021-01-25 16:16:26.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"07a5ee7837f9c243892ee383bb8cdcfbe0fe8d16","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-01-25 16:18:42.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)\n\nIt would be good to make sure we have some consensus from other Zuul maintainers before adding a new project in that namespace.","accounts_in_message":[],"_revision_number":2},{"id":"6bc219bd07d143f6ea27927ec701a6e805009671","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 16:24:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f30d266b600d4025be95c0e76a0b5b5f : SUCCESS in 6m 38s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/91b818c913764baca202212900d387e8 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":2},{"id":"5ca0ed446a1384c303ee63a40d1c5fedd9a60156","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2021-01-25 16:26:41.000000000","message":"Patch Set 2: Code-Review-1\n\nI think we may want this, but I\u0027d like some discussion on it first.\n\nCan you start a thread on the zuul-discuss mailing list so we can talk about the scope?  We may want this to be more than just swift.  And we probably want a different name.","accounts_in_message":[],"_revision_number":2},{"id":"2fa2e0e5a611a6d0780e418f7daff8c0c4e5ee1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2021-01-25 18:19:35.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"94bcd95d632cd07740597d38664d915110487649","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 18:29:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/bf1e12bf808f4090ae7e2f9f2d663a24 : SUCCESS in 5m 19s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/4d640316d1c14b1fa83968d94728b2b0 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":3},{"id":"efdaffc035dc0bb23411f2fccf46296931867bc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2021-01-26 13:54:01.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"865f74158e92ea27a88243ee2c1d678be4189d81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 14:05:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/fef4cf8b80af4b7da5430848f011348e : SUCCESS in 6m 02s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/31b0bc6aff2e4a64bcfe93b72f020842 : FAILURE in 3m 05s","accounts_in_message":[],"_revision_number":4},{"id":"940d831423df134cdf382eb945d88e9b0e9f8949","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2021-02-02 23:21:09.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM when it\u0027s in alphabetical order.","accounts_in_message":[],"_revision_number":4},{"id":"0cbfeb7f95b011793dd23c81eac48692cadd79f7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},"date":"2021-02-09 20:25:17.000000000","message":"Topic set to Open10K8S","accounts_in_message":[],"_revision_number":4},{"id":"19589f7ce4fab294ac4a044dc58d81dd0ef6427d","author":{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},"date":"2021-02-09 20:25:39.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"df8efd94918661ed710cfff9844cd6911af6fef1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2021-02-09 20:35:51.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"163aeb794beeeaed23d5f04c0691c94ec7f91452","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 20:47:26.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b1471c642b1a40afa050fd36b5b81df5 : SUCCESS in 6m 41s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/19fb538d06d84159a9a4858c3a6865c3 : FAILURE in 3m 37s","accounts_in_message":[],"_revision_number":5},{"id":"5cd1ceaefa84f0dbba2d7e48c1439a2eecd12b7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2021-02-10 11:47:59.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"182daa3c98699cddd6672578e224442543d9477b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 11:55:36.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e8eb2ef017174c7cbc7b8c19e8648894 : SUCCESS in 5m 15s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/39d8aad718844b5cb7cc4849963c35e0 : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":6},{"id":"ace7f3601c6fdea9dad2ea2b429740a7249bb95e","author":{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},"date":"2021-02-10 12:03:15.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"79062a83541444813be38122c93153ce3fc7e9cd","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-02-10 13:44:02.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"8f9585be8af482902c4945895a1e2eb66972545e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-02-10 15:27:07.000000000","message":"Patch Set 6:\n\nIs there any code to import or do you start from scratch with an empty repository?","accounts_in_message":[],"_revision_number":6},{"id":"a354304f9330491bab59ecb4e5c42f4d0734eac4","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2021-02-10 15:29:00.000000000","message":"Patch Set 6:\n\nI am planning to use https://review.opendev.org/c/zuul/zuul/+/708233 as the starting point.","accounts_in_message":[],"_revision_number":6},{"id":"7445275bfb36554430c51849b45e77fe9b97c4c3","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-02-10 15:53:58.000000000","message":"Patch Set 6:\n\nSo, you plan to go through a normal review process? Ok!","accounts_in_message":[],"_revision_number":6},{"id":"380f7b7059ae56e0063291439e10ee6e276950b9","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-02-10 15:56:36.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)\n\nmissing storyboard setup","accounts_in_message":[],"_revision_number":6},{"id":"e37e8ce0c7c993b9ae1838aab8734eabee4599fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2021-02-10 16:17:18.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"5f1debea8f7e847989c7dcd2c92ec91c4665a6ea","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-02-10 16:18:12.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"abfedc8ad1685a468ae6b68ffbe9329e5621a150","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-02-10 16:26:33.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"c7f5500291ac769b257a01df85823ef8cf1f2e00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 16:28:45.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/69d38017a8f143ef98a49cb9720d32cf : SUCCESS in 7m 38s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/e2aae317d389484b975cbb0b5f793958 : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":7},{"id":"25e0aee7ece80e1b56b51ecd79388be0f567aa3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 16:29:04.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"7834ad6f6a5e30d501549b1620d4a36c57decfb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 16:40:01.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3d80573037844cc2a348565f9cfe71cb : SUCCESS in 5m 32s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/88d732b9c6814ac3a3b1cd194fad2c8d : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":7},{"id":"094f05544e4c944154d35c110ae1c1a2eaac5f1e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 16:40:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"e1fbbc5d425fb90b18bc966f61e8dd391f80890f","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 17:37:01.000000000","message":"Patch Set 7:\n\nBuild failed (deploy pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/dc6b90c09fc8453e957074dfe5b7658e : TIMED_OUT in 30m 30s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"14a12f9dc6c46a23b5a1122efa345f2e70dc9bca","revisions":{"31d0109c9efd6d9e0cca74d894cfd888f69723bb":{"kind":"REWORK","_number":1,"created":"2021-01-25 15:42:21.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/64/772364/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/64/772364/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/772364/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/772364/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/772364/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/64/772364/1"}}},"commit":{"parents":[{"commit":"27ffaf3b44b28928348993b8763605125eb0aee8","subject":"Merge \"Set up access for #openinfra channel\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/27ffaf3b44b28928348993b8763605125eb0aee8"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-01-25 15:41:41.000000000","tz":120},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-01-25 15:41:41.000000000","tz":120},"subject":"Add swift-proxy in vexxhost tenant","message":"Add swift-proxy in vexxhost tenant\n\nref: https://review.opendev.org/c/openstack/project-config/+/771954\nChange-Id: Ic9e0f2071ac84deca60223471a35c087dde852a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/31d0109c9efd6d9e0cca74d894cfd888f69723bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/31d0109c9efd6d9e0cca74d894cfd888f69723bb"}]},"branch":"refs/heads/master"},"3867cc44cfba2fb5ee9a7568a9b4c4f9a899597a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-01-25 16:16:26.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/64/772364/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/64/772364/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/772364/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/772364/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/772364/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/64/772364/2"}}},"commit":{"parents":[{"commit":"27ffaf3b44b28928348993b8763605125eb0aee8","subject":"Merge \"Set up access for #openinfra channel\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/27ffaf3b44b28928348993b8763605125eb0aee8"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-01-25 15:41:41.000000000","tz":120},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-01-25 16:15:53.000000000","tz":120},"subject":"Add swift-proxy in vexxhost tenant","message":"Add swift-proxy in vexxhost tenant\n\nref: https://review.opendev.org/c/openstack/project-config/+/772331\nChange-Id: Ic9e0f2071ac84deca60223471a35c087dde852a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3867cc44cfba2fb5ee9a7568a9b4c4f9a899597a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3867cc44cfba2fb5ee9a7568a9b4c4f9a899597a"}]},"branch":"refs/heads/master"},"53dde62ae422bc8527c8ac0c47ea131aa818ca4e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-01-25 18:19:35.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/64/772364/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/64/772364/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/772364/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/772364/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/772364/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/64/772364/3"}}},"commit":{"parents":[{"commit":"27ffaf3b44b28928348993b8763605125eb0aee8","subject":"Merge \"Set up access for #openinfra channel\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/27ffaf3b44b28928348993b8763605125eb0aee8"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-01-25 15:41:41.000000000","tz":120},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-01-25 18:19:02.000000000","tz":120},"subject":"Add swift-proxy in zuul namespace","message":"Add swift-proxy in zuul namespace\n\nref: https://review.opendev.org/c/openstack/project-config/+/772331\nChange-Id: Ic9e0f2071ac84deca60223471a35c087dde852a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/53dde62ae422bc8527c8ac0c47ea131aa818ca4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/53dde62ae422bc8527c8ac0c47ea131aa818ca4e"}]},"branch":"refs/heads/master"},"8a0894c7ee30d0f76e92654a5cdb5e282955c481":{"kind":"REWORK","_number":4,"created":"2021-01-26 13:54:01.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/64/772364/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/64/772364/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/772364/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/772364/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/772364/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/64/772364/4"}}},"commit":{"parents":[{"commit":"27ffaf3b44b28928348993b8763605125eb0aee8","subject":"Merge \"Set up access for #openinfra channel\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/27ffaf3b44b28928348993b8763605125eb0aee8"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-01-25 15:41:41.000000000","tz":120},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-01-26 13:47:19.000000000","tz":120},"subject":"Add zuul-storage-proxy in zuul namespace","message":"Add zuul-storage-proxy in zuul namespace\n\nref: https://review.opendev.org/c/openstack/project-config/+/772331\nChange-Id: Ic9e0f2071ac84deca60223471a35c087dde852a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8a0894c7ee30d0f76e92654a5cdb5e282955c481"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8a0894c7ee30d0f76e92654a5cdb5e282955c481"}]},"branch":"refs/heads/master"},"176ed6a22b2ed7ad07116dc28d6729863a99aaef":{"kind":"REWORK","_number":5,"created":"2021-02-09 20:35:51.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/64/772364/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/64/772364/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/772364/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/772364/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/772364/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/64/772364/5"}}},"commit":{"parents":[{"commit":"27ffaf3b44b28928348993b8763605125eb0aee8","subject":"Merge \"Set up access for #openinfra channel\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/27ffaf3b44b28928348993b8763605125eb0aee8"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-01-25 15:41:41.000000000","tz":120},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-02-09 20:35:07.000000000","tz":120},"subject":"Add zuul-storage-proxy in zuul namespace","message":"Add zuul-storage-proxy in zuul namespace\n\nref: https://review.opendev.org/c/zuul/zuul/+/708233\nChange-Id: Ic9e0f2071ac84deca60223471a35c087dde852a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/176ed6a22b2ed7ad07116dc28d6729863a99aaef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/176ed6a22b2ed7ad07116dc28d6729863a99aaef"}]},"branch":"refs/heads/master"},"32d39933991e299bba0d979e3215442c27ecc9eb":{"kind":"REWORK","_number":6,"created":"2021-02-10 11:47:59.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/64/772364/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/64/772364/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/772364/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/772364/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/772364/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/64/772364/6"}}},"commit":{"parents":[{"commit":"27ffaf3b44b28928348993b8763605125eb0aee8","subject":"Merge \"Set up access for #openinfra channel\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/27ffaf3b44b28928348993b8763605125eb0aee8"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-01-25 15:41:41.000000000","tz":120},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-02-10 11:27:35.000000000","tz":120},"subject":"Add zuul-storage-proxy in zuul namespace","message":"Add zuul-storage-proxy in zuul namespace\n\nref: https://review.opendev.org/c/zuul/zuul/+/708233\nChange-Id: Ic9e0f2071ac84deca60223471a35c087dde852a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/32d39933991e299bba0d979e3215442c27ecc9eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/32d39933991e299bba0d979e3215442c27ecc9eb"}]},"branch":"refs/heads/master"},"14a12f9dc6c46a23b5a1122efa345f2e70dc9bca":{"kind":"REWORK","_number":7,"created":"2021-02-10 16:17:18.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/64/772364/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/64/772364/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/772364/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/772364/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/772364/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/64/772364/7"}}},"commit":{"parents":[{"commit":"27ffaf3b44b28928348993b8763605125eb0aee8","subject":"Merge \"Set up access for #openinfra channel\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/27ffaf3b44b28928348993b8763605125eb0aee8"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-01-25 15:41:41.000000000","tz":120},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-02-10 16:10:03.000000000","tz":120},"subject":"Add zuul-storage-proxy in zuul namespace","message":"Add zuul-storage-proxy in zuul namespace\n\nref: https://review.opendev.org/c/zuul/zuul/+/708233\nChange-Id: Ic9e0f2071ac84deca60223471a35c087dde852a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/14a12f9dc6c46a23b5a1122efa345f2e70dc9bca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/14a12f9dc6c46a23b5a1122efa345f2e70dc9bca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
