)]}'
{"id":"openstack%2Fproject-config~772427","triplet_id":"openstack%2Fproject-config~master~I736c807030cffff8d9017cbbbc8a252de902d423","project":"openstack/project-config","branch":"master","topic":"allow_edit_hashtag","hashtags":[],"change_id":"I736c807030cffff8d9017cbbbc8a252de902d423","subject":"Update ACLs of Ironic Projects to allow Edit Hashtags","status":"MERGED","created":"2021-01-25 19:53:14.000000000","updated":"2021-02-03 18:40:33.000000000","submitted":"2021-02-03 18:10:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772427-allow_edit_hashtag","meta_rev_id":"3a17b606cffcbe68a736ff99612487c9c0586ae8","_number":772427,"virtual_id_number":772427,"owner":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-03 18:10:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2021-02-03 18:04:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2021-02-03 11:25:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":1,"date":"2021-02-01 15:06:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-01 05:25:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":1,"date":"2021-02-02 14:45:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2021-02-03 18:04:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-25 20:18:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-26 09:25:46.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"CC"},{"updated":"2021-01-29 06:26:46.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2021-02-01 15:06:43.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2021-02-02 14:45:49.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2021-02-03 11:25:12.000000000","updated_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"reviewer":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"state":"REVIEWER"},{"updated":"2021-02-03 18:04:23.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"0cb1be7be6ee81f4429a1a8c0e81d99f578ef8ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-01-25 19:53:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"077e8682ff5c91b8e20d1df723a28c0250fc8f9b","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-01-25 19:59:45.000000000","message":"Patch Set 1:\n\nLet\u0027s see how it goes \u003d) I\u0027m not 100% sure about the syntax","accounts_in_message":[],"_revision_number":1},{"id":"f3cda8766e9ad3d1f37933aa110c56bd59e92253","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 20:18:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/dd41cdf2ffc04877a6d119f7d08b65b0 : SUCCESS in 6m 27s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/ef43bd7963784fa08795eb0cba2fc2a4 : FAILURE in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"76425aa8c0fe629a48eb1025216d0c1fb99ae3ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-01-25 21:13:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"29c1893dcc530d5b12014f19cacaabc480540218","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 21:26:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f5188faf60534f71b9c88176c92e7c82 : SUCCESS in 7m 39s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/83f9889003db4046a5449c295681d73e : FAILURE in 4m 44s","accounts_in_message":[],"_revision_number":2},{"id":"29a9765873012e07eafde35b66db53eaf51134c4","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-01-26 09:25:46.000000000","message":"Patch Set 2:\n\nthis https://review.opendev.org/c/openstack/project-config/+/772484 will probably help :)","accounts_in_message":[],"_revision_number":2},{"id":"399deed1b1b68e02a9393441cd6a7d02be87df13","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-01-26 12:30:01.000000000","message":"Patch Set 2:\n\nTks Riccardo!","accounts_in_message":[],"_revision_number":2},{"id":"598c6a620cf7c552cc243e921bf727502258b4a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-01-27 17:10:18.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"5c4d931adb8724ece115676cf6b093685dde8e21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 17:18:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/55b682ac212c474fbb6e67599ee05061 : SUCCESS in 5m 36s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/7201b29c20df48f5a712166d0eb0e6cf : FAILURE in 3m 27s","accounts_in_message":[],"_revision_number":3},{"id":"c9e6e56bea782c430483307572bf740a055ce6a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-01-27 18:35:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4490f171f2cac445126528f596ad5089ade34e97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 18:45:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1c5bd6452bfb408cb4fe3b640216d4f5 : SUCCESS in 4m 52s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/4fe7e1273b6a4f1e9f483ce1230d474a : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":4},{"id":"ca0ebb00c73d27aa3b2130c36d282d5ff9658151","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-01-29 06:26:46.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c601b86672dfed4ce070a18f3c65a3771eea623f","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-01-29 06:27:37.000000000","message":"Patch Set 4: Code-Review+1\n\nit might be worth discussing if this is just something that should be globally allowed for everyone.  does it need to be a core only thing?","accounts_in_message":[],"_revision_number":4},{"id":"6c919ff3c858e27324531cdfafbbf4176aa7ef21","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-01-29 10:53:07.000000000","message":"Patch Set 4:\n\nHi Ian,\n\nSince the default for edit hashtag is only for the owner/uploader of the patch, our idea in the Ironic community s to allow cores to edit so we can keep track of priorities and patches we would like to backport for things.","accounts_in_message":[],"_revision_number":4},{"id":"e34cdaddd6b7b7528a90681f8fe79f4964901750","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-02-01 05:25:52.000000000","message":"Patch Set 4: Code-Review+2\n\nIt seems reasonable","accounts_in_message":[],"_revision_number":4},{"id":"93876531e6bd0173ae68c25371235b94f691b5f6","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-02-01 15:06:43.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"6d41ed098ade72f89afb5fbec843b834cf8c9f5f","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-02-02 14:45:49.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"ee31ccd9e9454158576ab5cebe247ed1da5ad00d","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2021-02-03 11:25:12.000000000","message":"Patch Set 4: Code-Review+1\n\nThank you Iury.","accounts_in_message":[],"_revision_number":4},{"id":"50aab70826d5b834e3f0113094402cd158e45545","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-02-03 18:04:23.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"afba968a61fff52f01c92c3fbedc7d252531ecb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 18:04:45.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"995fd94ab45e8cea4d03173d526dec518c3f42e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 18:10:36.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5e6d1d8a472f46a590c8e3a11739a3b0 : SUCCESS in 4m 48s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f51a200ef1f04d40937741ef9b14ddf0 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":4},{"id":"237b350a750ef0685f19a0c3f9107f27213b8627","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 18:10:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"3a17b606cffcbe68a736ff99612487c9c0586ae8","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 18:40:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/7298be7458df4e70b23af882ae32dd38 : SUCCESS in 17m 01s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"dd2ae03e6c55d968a0550054ea033dcf0bc77c0b","revisions":{"7ba53fa78ef828b0858f09520e13d7b0bc9d10d9":{"kind":"REWORK","_number":1,"created":"2021-01-25 19:53:14.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/27/772427/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/27/772427/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/772427/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/772427/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/772427/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/27/772427/1"}}},"commit":{"parents":[{"commit":"7181a05d964d6b323107c11bcd76e76aae44164c","subject":"Revert \"Revert \"Revert \"Revert \"Temporarily stop booting nodes in inap-mtl01\"\"\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7181a05d964d6b323107c11bcd76e76aae44164c"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2021-01-25 19:48:58.000000000","tz":60},"committer":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2021-01-25 19:53:11.000000000","tz":60},"subject":"Update ACLs of Ironic Projects to allow Edit Hashtags","message":"Update ACLs of Ironic Projects to allow Edit Hashtags\n\nThis patch aims to update the ACL for some of the Ironic\nprojecsts to allow the edit of hashtags\n\nProjects updated:\n- bifrost\n- ironic-inspector\n- ironic-specs\n- ironic-ui\n- ironic\n- metalsmith\n- networking-generic-switch\n- tenks\n- virtualbmc\n- ironic-staging-drivers\n\nChange-Id: I736c807030cffff8d9017cbbbc8a252de902d423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7ba53fa78ef828b0858f09520e13d7b0bc9d10d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7ba53fa78ef828b0858f09520e13d7b0bc9d10d9"}]},"branch":"refs/heads/master"},"ab16d77258117fd00dbf64ec2eb38ba57cbb0515":{"kind":"REWORK","_number":2,"created":"2021-01-25 21:13:26.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/27/772427/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/27/772427/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/772427/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/772427/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/772427/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/27/772427/2"}}},"commit":{"parents":[{"commit":"7181a05d964d6b323107c11bcd76e76aae44164c","subject":"Revert \"Revert \"Revert \"Revert \"Temporarily stop booting nodes in inap-mtl01\"\"\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7181a05d964d6b323107c11bcd76e76aae44164c"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2021-01-25 19:48:58.000000000","tz":60},"committer":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2021-01-25 21:13:17.000000000","tz":60},"subject":"Update ACLs of Ironic Projects to allow Edit Hashtags","message":"Update ACLs of Ironic Projects to allow Edit Hashtags\n\nThis patch aims to update the ACL for some of the Ironic\nprojecsts to allow the edit of hashtags\n\nProjects updated:\n- bifrost\n- ironic-inspector\n- ironic-specs\n- ironic-ui\n- ironic\n- metalsmith\n- networking-generic-switch\n- tenks\n- virtualbmc\n- ironic-staging-drivers\n\nChange-Id: I736c807030cffff8d9017cbbbc8a252de902d423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ab16d77258117fd00dbf64ec2eb38ba57cbb0515"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ab16d77258117fd00dbf64ec2eb38ba57cbb0515"}]},"branch":"refs/heads/master"},"9c45a3f00164c477783a1d3c1ae7c3bb0108266a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-01-27 17:10:18.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/27/772427/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/27/772427/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/772427/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/772427/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/772427/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/27/772427/3"}}},"commit":{"parents":[{"commit":"c970a43d470b837f62a6fd377b5464117ca41c5c","subject":"Add key editHashtags to normalize_acl script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c970a43d470b837f62a6fd377b5464117ca41c5c"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2021-01-25 19:48:58.000000000","tz":60},"committer":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2021-01-27 17:10:07.000000000","tz":60},"subject":"Update ACLs of Ironic Projects to allow Edit Hashtags","message":"Update ACLs of Ironic Projects to allow Edit Hashtags\n\nThis patch aims to update the ACL for some of the Ironic\nprojecsts to allow the edit of hashtags\n\nProjects updated:\n- bifrost\n- ironic-inspector\n- ironic-specs\n- ironic-ui\n- ironic\n- metalsmith\n- networking-generic-switch\n- tenks\n- virtualbmc\n- ironic-staging-drivers\n\nChange-Id: I736c807030cffff8d9017cbbbc8a252de902d423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9c45a3f00164c477783a1d3c1ae7c3bb0108266a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9c45a3f00164c477783a1d3c1ae7c3bb0108266a"}]},"branch":"refs/heads/master"},"dd2ae03e6c55d968a0550054ea033dcf0bc77c0b":{"kind":"REWORK","_number":4,"created":"2021-01-27 18:35:46.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/27/772427/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/27/772427/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/772427/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/772427/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/772427/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/27/772427/4"}}},"commit":{"parents":[{"commit":"0a3e98f7b8c92c2180287632d789bb5ad55b36c7","subject":"Add key editHashtags to normalize_acl script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0a3e98f7b8c92c2180287632d789bb5ad55b36c7"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2021-01-25 19:48:58.000000000","tz":60},"committer":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2021-01-27 18:31:50.000000000","tz":60},"subject":"Update ACLs of Ironic Projects to allow Edit Hashtags","message":"Update ACLs of Ironic Projects to allow Edit Hashtags\n\nThis patch aims to update the ACL for some of the Ironic\nprojecsts to allow the edit of hashtags\n\nProjects updated:\n- bifrost\n- ironic-inspector\n- ironic-specs\n- ironic-ui\n- ironic\n- metalsmith\n- networking-generic-switch\n- sushy\n- tenks\n- virtualbmc\n\nChange-Id: I736c807030cffff8d9017cbbbc8a252de902d423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dd2ae03e6c55d968a0550054ea033dcf0bc77c0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dd2ae03e6c55d968a0550054ea033dcf0bc77c0b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
