)]}'
{"id":"openstack%2Fproject-config~773240","triplet_id":"openstack%2Fproject-config~master~I72c8344d60a3f6fe184b1ffc8a64c8f8a8d78c80","project":"openstack/project-config","branch":"master","topic":"citycloud-nodes","hashtags":[],"change_id":"I72c8344d60a3f6fe184b1ffc8a64c8f8a8d78c80","subject":"Revert \"Temporarily stop booting nodes in citycloud-kna1\"","status":"MERGED","created":"2021-01-30 14:37:59.000000000","updated":"2021-02-02 19:48:03.000000000","submitted":"2021-02-02 19:40:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773240-citycloud-nodes","meta_rev_id":"2b66ef713309c07e6c9640c8cbee97fc14f391d2","_number":773240,"virtual_id_number":773240,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-02 19:40:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-02 19:07:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2021-02-02 19:07:11.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":1,"date":"2021-02-02 19:30:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-30 14:42:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-01 00:16:21.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"}],"messages":[{"id":"f82883bc23f232cb50dd26af3549051519500c02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-01-30 14:37:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"697e37fbf91d0da829828e8dcc1e16833da5dd5b","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-01-30 14:39:56.000000000","message":"Patch Set 1: Workflow-1\n\nWIP until things are back in shape here.","accounts_in_message":[],"_revision_number":1},{"id":"9f18030f4c0b7ba5eb6d1483c13284b812ce55cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-30 14:42:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a7894db2edb640d3b0af8cc0bccf7f55 : SUCCESS in 4m 15s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/38bc9f515c594237af639ffac8be9dd4 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":1},{"id":"92aec136c977fb405b15a5afdd4d10bfee959f15","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-01-31 23:10:09.000000000","message":"Patch Set 1: -Workflow\n\nRemoving WIP on this to simplify approval in case someone arrives at a fix, but don\u0027t approve unless the mirror is actually back to operating correctly again.","accounts_in_message":[],"_revision_number":1},{"id":"afc07d2323dff8ab9728190b64f83d7caf9a744c","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-02-01 00:16:21.000000000","message":"Patch Set 1: Workflow-1\n\nnote as part of testing I have changed the floating ip; so we will need to update DNS first.  At this point it seems to be an upstream networking error, so we are pending investigation from the cloud provider at this point.  I\u0027ll re-mark it WIP and watch for updates","accounts_in_message":[],"_revision_number":1},{"id":"c238b96e08dfb4ab4d722450986474c6512a23c4","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-02-02 18:39:03.000000000","message":"Patch Set 1:\n\nDNS has been updated with I898dbf7417fb01f608eded85faaae5a417ad2e98 for longer than the hour-long TTL.","accounts_in_message":[],"_revision_number":1},{"id":"e8843f2f0d3b8ec7f657d494cb85da315392efcb","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-02-02 19:04:03.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"b1670627478129465f14edb0a963dd670453d5ab","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-02-02 19:07:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"30695706f8f98fc106d97c979ffec532bfa7dbfd","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-02-02 19:30:08.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8139e3dc1655641f3db89268edb2b7d7c66893e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 19:30:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"13bfb466e2bf634f78f9b98e0f72a8edb810355d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 19:40:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/46c0f04fb8514fbc9e4b6340ed4a2c2a : SUCCESS in 5m 19s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/ab52514e08cb4f5fa28e53c99fc449dc : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":1},{"id":"cf01e082a604345bfd06664d22e5c86f962c3864","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 19:40:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2b66ef713309c07e6c9640c8cbee97fc14f391d2","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 19:48:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/d7e760f636874fdba92a79f553c03c92 : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"379eb2869d1c48813e2e1ad62c7e49f7e71ddd78","revisions":{"379eb2869d1c48813e2e1ad62c7e49f7e71ddd78":{"kind":"REWORK","_number":1,"created":"2021-01-30 14:37:59.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/40/773240/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/40/773240/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/773240/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/773240/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/773240/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/40/773240/1"}}},"commit":{"parents":[{"commit":"a7edbb7a18ad5e48dcc025dba6583cf13284f87a","subject":"Temporarily stop booting nodes in citycloud-kna1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a7edbb7a18ad5e48dcc025dba6583cf13284f87a"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-01-30 14:37:11.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-01-30 14:37:18.000000000","tz":0},"subject":"Revert \"Temporarily stop booting nodes in citycloud-kna1\"","message":"Revert \"Temporarily stop booting nodes in citycloud-kna1\"\n\nThis reverts commit a7edbb7a18ad5e48dcc025dba6583cf13284f87a.\n\nMerge once the mirror server in citycloud-kna1 is reachable and\nworking correctly.\n\nChange-Id: I72c8344d60a3f6fe184b1ffc8a64c8f8a8d78c80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/379eb2869d1c48813e2e1ad62c7e49f7e71ddd78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/379eb2869d1c48813e2e1ad62c7e49f7e71ddd78"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
