)]}'
{"id":"openstack%2Fproject-config~773710","triplet_id":"openstack%2Fproject-config~master~Ie424e7d6236c00678d92b34ed33dafaf966180fa","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"Ie424e7d6236c00678d92b34ed33dafaf966180fa","subject":"Switch to using v3-standard-8 flavors","status":"MERGED","created":"2021-02-02 15:33:30.000000000","updated":"2021-02-02 16:03:44.000000000","submitted":"2021-02-02 15:56:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773710","meta_rev_id":"a4eaabab710684a33821ed81cd9b7d9374df1d65","_number":773710,"virtual_id_number":773710,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-02 15:56:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2021-02-02 15:42:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-02 15:39:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-02 15:45:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-02 15:39:45.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2021-02-02 15:41:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-02 15:42:26.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"7d22c6db02f452ed7d010135d397a683f615f347","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2021-02-02 15:33:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ecfea2f78797cfe8558745677ad5aedde714e69","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-02-02 15:39:45.000000000","message":"Patch Set 1: Code-Review+2\n\nIf that\u0027s what\u0027s available then we\u0027ll need to use it. We may however need to revisit our old kernel parameters for artificially limiting system memory at boot time, because otherwise it will be all too easy for a project to merge a change which triples their memory utilization just because they happened to run on nodes with 32GB RAM instead of 8GB.","accounts_in_message":[],"_revision_number":1},{"id":"00dbde680255f250f4a7d60100e5451230eeb956","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 15:41:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/518e5b827b1a45f49f3c97a30029b5e8 : SUCCESS in 5m 09s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/67d80d8e24ef4861b827c2f661c644f0 : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":1},{"id":"d50a4e2481b57628e8c3bbafa1c1ae3e8cbacf26","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-02-02 15:42:26.000000000","message":"Patch Set 1: Code-Review+2\n\nThere may be some objection to this for making our resources differ too much between providers, but I\u0027m thinking having some \"natural\" flavor size makes sense and we should check with other clouds whether we can do similar adjustments there.","accounts_in_message":[],"_revision_number":1},{"id":"d8f54b61c6061b3882e41606d0ada27d6e594e97","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-02-02 15:45:51.000000000","message":"Patch Set 1: Workflow+1\n\nGoing ahead and approving this now, but we should bring it up in today\u0027s meeting and/or on the ML so folks are aware, and we can adjust our configuration or user expectations accordingly.","accounts_in_message":[],"_revision_number":1},{"id":"d9f821f25eb6642115e386f321458a7ec3c9c03e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 15:46:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ea9769bcbd34d647554b6279bf4e9583f5971faf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 15:56:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b5c0bf25cca9407fa226bed6f7f56942 : SUCCESS in 5m 42s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/e06f43fc18ad451bbe1a20397731ca59 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"d8570b161f7585e419cc222dbefabf4494295fe3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 15:56:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a4eaabab710684a33821ed81cd9b7d9374df1d65","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 16:03:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/d3c1d9f2e2ca4908a2115b86a17a0941 : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2a28bb03ee314c414a2211ae77cbdee1247ffd12","revisions":{"2a28bb03ee314c414a2211ae77cbdee1247ffd12":{"kind":"REWORK","_number":1,"created":"2021-02-02 15:33:30.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/10/773710/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/10/773710/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/773710/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/773710/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/773710/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/10/773710/1"}}},"commit":{"parents":[{"commit":"57ed60678ea1309ae8c580364aa25baeb073983d","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/57ed60678ea1309ae8c580364aa25baeb073983d"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-02-02 15:30:31.000000000","tz":-300},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-02-02 15:33:29.000000000","tz":-300},"subject":"Switch to using v3-standard-8 flavors","message":"Switch to using v3-standard-8 flavors\n\nVEXXHOST no longer offers to flavors of instances, and it makes more\nsense to use the standard flavors.  This flavor has 8 cores, 32GB\nof memory and 160GB of disk space.  It doesn\u0027t make sense to have\na smaller memory/disk allocation as the hypervisors are ratio\u0027d out\nto work with this configuration.\n\nChange-Id: Ie424e7d6236c00678d92b34ed33dafaf966180fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2a28bb03ee314c414a2211ae77cbdee1247ffd12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2a28bb03ee314c414a2211ae77cbdee1247ffd12"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
