)]}'
{"id":"openstack%2Fproject-config~773936","triplet_id":"openstack%2Fproject-config~master~I9c75e12d6b9c2b1c7c4a3047a90e9f80440c3532","project":"openstack/project-config","branch":"master","topic":"new-project","hashtags":[],"change_id":"I9c75e12d6b9c2b1c7c4a3047a90e9f80440c3532","subject":"New Project Request: airship/gerrit-to-github-bot","status":"MERGED","created":"2021-02-03 16:19:29.000000000","updated":"2021-02-03 22:58:30.000000000","submitted":"2021-02-03 19:49:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"773936-new-project","meta_rev_id":"059ece87d476b121f0ff2aca289ec7aaba89d7cb","_number":773936,"virtual_id_number":773936,"owner":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-03 19:49:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-03 19:35:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-03 19:35:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-03 16:23:43.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2021-02-03 16:36:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-03 19:35:50.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"6446ee192dfb770eac94eb6ae35b2d179087bedd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"date":"2021-02-03 16:19:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73f76bfa7a96e2dc24ea6815f522ba35325d0e3e","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-02-03 16:23:43.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dec820e620996b3a27edfa1db7aa938a55a813dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 16:36:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/bebe8710c1634506a44f7931ffa74f27 : SUCCESS in 4m 27s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b29bdf78786d4af994e7e435c8565956 : FAILURE in 4m 10s","accounts_in_message":[],"_revision_number":1},{"id":"0b2c6e2eb5d9acbf70c59bbba40ea180307df157","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"date":"2021-02-03 19:29:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ccc96cb65c5283a495e6043c53d2184945a5ff9e","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-02-03 19:35:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"19b8136bfc3405bbfbdac43918a0af72ed394782","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 19:39:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/05002e08a3194f838de1db9e39de35cf : SUCCESS in 5m 15s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/620364c0204440ecbde0236442d1c3e3 : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":2},{"id":"f7203feda38c0bad09fad6276af20d1b1e20cf5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 19:40:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"540aab04cfc63ba34c16bcb6b463f9a4b28cbff1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 19:49:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/07e9e8bc165a4c6db954959e6fc9b4ac : SUCCESS in 5m 07s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/dc366c8df0c14b0e95205fc175d3cc8a : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":2},{"id":"fb1c08ca5f82565999566c78a6210fd00f35ab87","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 19:49:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"982d32ae58d17792281dc4e1c7a6e60af7d7460d","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 20:23:30.000000000","message":"Patch Set 2:\n\nBuild failed (deploy pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/5807ab7e17e24d1f9cb76c1297936a39 : TIMED_OUT in 30m 28s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":2},{"id":"059ece87d476b121f0ff2aca289ec7aaba89d7cb","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-02-03 22:58:30.000000000","message":"Patch Set 2:\n\nMatt: I\u0027ve added you as the initial member of gerrit-to-github-bot-core just now. Enjoy!","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"97d53affe467e16ea8c297fd5a9b968d078ad025","revisions":{"60f678dc747d8b37fde956a6a698e34fa9ea5089":{"kind":"REWORK","_number":1,"created":"2021-02-03 16:19:29.000000000","uploader":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"ref":"refs/changes/36/773936/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/36/773936/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/36/773936/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/36/773936/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/36/773936/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/36/773936/1"}}},"commit":{"parents":[{"commit":"049692fb63f80b58741d2727ebb1f81e955857a5","subject":"Merge \"CentOS 8 Stream initial enablement for AArch64\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/049692fb63f80b58741d2727ebb1f81e955857a5"}]}],"author":{"name":"Matt McEuen","email":"madgin@madgin.net","date":"2021-02-03 16:15:44.000000000","tz":-360},"committer":{"name":"Matt McEuen","email":"madgin@madgin.net","date":"2021-02-03 16:15:44.000000000","tz":-360},"subject":"New Project Request: airship/gerrit-to-github-bot","message":"New Project Request: airship/gerrit-to-github-bot\n\nThis is a request to create the airship/gerrit-to-github-bot\nproject in opendev.\n\nThis bot synchronizes state between gerrit patchsets, where\nairship work is done, and github issues, where airship scope\nis tracked.\n\nIf you can seed the new gerrit group with me then I can add\nthe appropriate core team.\n\nThe upstream imported project, while created for Airship in\nthe first place, is essentially abandoned by its owner (hence our\nneed to import to airship proper to continue enhancing it).\nWe are unable to clean up the extra feature branch outside of master.\n\nChange-Id: I9c75e12d6b9c2b1c7c4a3047a90e9f80440c3532\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/60f678dc747d8b37fde956a6a698e34fa9ea5089"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/60f678dc747d8b37fde956a6a698e34fa9ea5089"}]},"branch":"refs/heads/master"},"97d53affe467e16ea8c297fd5a9b968d078ad025":{"kind":"REWORK","_number":2,"created":"2021-02-03 19:29:10.000000000","uploader":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"ref":"refs/changes/36/773936/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/36/773936/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/36/773936/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/36/773936/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/36/773936/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/36/773936/2"}}},"commit":{"parents":[{"commit":"049692fb63f80b58741d2727ebb1f81e955857a5","subject":"Merge \"CentOS 8 Stream initial enablement for AArch64\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/049692fb63f80b58741d2727ebb1f81e955857a5"}]}],"author":{"name":"Matt McEuen","email":"madgin@madgin.net","date":"2021-02-03 16:15:44.000000000","tz":-360},"committer":{"name":"Matt McEuen","email":"madgin@madgin.net","date":"2021-02-03 19:28:59.000000000","tz":-360},"subject":"New Project Request: airship/gerrit-to-github-bot","message":"New Project Request: airship/gerrit-to-github-bot\n\nThis is a request to create the airship/gerrit-to-github-bot\nproject in opendev.\n\nThis bot synchronizes state between gerrit patchsets, where\nairship work is done, and github issues, where airship scope\nis tracked.\n\nIf you can seed the new gerrit group with me then I can add\nthe appropriate core team.\n\nThe upstream imported project, while created for Airship in\nthe first place, is essentially abandoned by its owner (hence our\nneed to import to airship proper to continue enhancing it).\nWe are unable to clean up the extra feature branch outside of master.\n\nChange-Id: I9c75e12d6b9c2b1c7c4a3047a90e9f80440c3532\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/97d53affe467e16ea8c297fd5a9b968d078ad025"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/97d53affe467e16ea8c297fd5a9b968d078ad025"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
