)]}'
{"id":"openstack%2Fproject-config~774550","triplet_id":"openstack%2Fproject-config~master~I4d09979e0c9dac75c7f3d878cde99cdc56b8dbe3","project":"openstack/project-config","branch":"master","topic":"openifra-channels","hashtags":[],"change_id":"I4d09979e0c9dac75c7f3d878cde99cdc56b8dbe3","subject":"Setup OpenInfra Channels","status":"MERGED","created":"2021-02-08 22:11:06.000000000","updated":"2021-02-09 00:52:21.000000000","submitted":"2021-02-09 00:32:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774550-openifra-channels","meta_rev_id":"500e5d99ae73c881cc3d868980ffbf2027be8cb7","_number":774550,"virtual_id_number":774550,"owner":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-09 00:32:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-09 00:24:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-09 00:24:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-08 22:18:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-08 23:32:14.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2021-02-09 00:24:21.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"e4e1ac9bdd256c485b8235da431e8758e2636749","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2021-02-08 22:11:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7cab65f8617903f921e3091688d02440522f8181","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 22:18:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/03021262c9664dbab1388232a83f809e : SUCCESS in 4m 31s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/778a66c4185340f193b66c1266d0f947 : FAILURE in 6m 35s","accounts_in_message":[],"_revision_number":1},{"id":"0466ed2fa953e4c1eb2f95a2782e0834406b1eca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2021-02-08 23:15:52.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"d6d7820c65e414bd25ead93b793ac630dcea320f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 23:23:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/17ef2213f21f40cdabc77c9e6e21a57a : SUCCESS in 5m 01s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/5ca6211c89fc47f882d3e8f3c724310f : FAILURE in 6m 21s","accounts_in_message":[],"_revision_number":2},{"id":"8dc02a9a6ea2b565aa8b2fdebcc47a1fa5ca997e","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-02-08 23:32:14.000000000","message":"Patch Set 2:\n\nrecheck since I see #openinfra-diversity as registered (lag between IRC servers maybe?)","accounts_in_message":[],"_revision_number":2},{"id":"c34cacc150b0cb2e6a7f209c7ac6bc32c39f5bd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 23:41:08.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/64700abecc914f468b1c65a6ad7eedbf : SUCCESS in 5m 23s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/ec26df8eff544deab0120478b93d38a2 : FAILURE in 7m 33s","accounts_in_message":[],"_revision_number":2},{"id":"97d6dedcda0b6916981452b081072bce4daeb89e","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-02-08 23:53:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"983dcdfb2ad4be30cad956224a849865bdeb5145","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2021-02-09 00:03:04.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"0bcbd3c6c278556b14815639a42f8d4c72b6a528","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 00:10:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8735e17caf5b4400a83afbaf37b8ed87 : SUCCESS in 6m 08s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/a0e7e54923b24ddcb84a696ffb5c26f6 : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":3},{"id":"57b8061b8d6a04cfc9b82e580987b40df31e110f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-02-09 00:24:21.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d50d8768d88e8840180223879081e76fce35e391","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 00:24:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5a71c22dacf4d6c5615289955b3c979e5e809f47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 00:32:06.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/fe5f422b212440b987767b48ef2daf2d : SUCCESS in 4m 52s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/7349655dd11a4cd68c0cd6a8a72c9fd8 : SUCCESS in 7m 17s","accounts_in_message":[],"_revision_number":3},{"id":"202dcf0a2f3e3c32dd14f2fed045dc9e5bb3a78d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 00:32:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"500e5d99ae73c881cc3d868980ffbf2027be8cb7","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 00:52:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-run-accessbot https://zuul.opendev.org/t/openstack/build/71f07a9fcdf440e39054166e6ff87748 : SUCCESS in 9m 47s\n- infra-prod-service-eavesdrop https://zuul.opendev.org/t/openstack/build/aa35e9daa89442b0a19de46126b256de : SUCCESS in 6m 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f9c31de4c134c880d7ef67c3f074dc4ffcf0a2b8","revisions":{"83d21ea538d6fd018a5b76e5d586756ee526cc6e":{"kind":"REWORK","_number":1,"created":"2021-02-08 22:11:06.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/50/774550/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/774550/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/774550/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/774550/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/774550/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/774550/1"}}},"commit":{"parents":[{"commit":"e707d93f2b38b5c1b0645fdc2f9e62951f6ded52","subject":"Merge \"zuul-worker: remove additional install of apt-transport-https\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e707d93f2b38b5c1b0645fdc2f9e62951f6ded52"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-02-08 22:06:02.000000000","tz":-480},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-02-08 22:06:02.000000000","tz":-480},"subject":"Setup OpenInfra Channels","message":"Setup OpenInfra Channels\n\nopenstack-forum, openstack-ptg, openstack-summit and openinfra-summit\nwill now be openinfra-events.\n\nopenstack-diversity will now be openinfra-diversity.\n\nopenstack-board will now be openinfra-board.\n\nopenstack-foundation will now be openinfra.\n\nChange-Id: I4d09979e0c9dac75c7f3d878cde99cdc56b8dbe3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/83d21ea538d6fd018a5b76e5d586756ee526cc6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/83d21ea538d6fd018a5b76e5d586756ee526cc6e"}]},"branch":"refs/heads/master"},"0610c41995ebdea4654703870b6db4617941de41":{"kind":"REWORK","_number":2,"created":"2021-02-08 23:15:52.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/50/774550/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/774550/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/774550/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/774550/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/774550/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/774550/2"}}},"commit":{"parents":[{"commit":"e707d93f2b38b5c1b0645fdc2f9e62951f6ded52","subject":"Merge \"zuul-worker: remove additional install of apt-transport-https\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e707d93f2b38b5c1b0645fdc2f9e62951f6ded52"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-02-08 22:06:02.000000000","tz":-480},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-02-08 23:15:45.000000000","tz":0},"subject":"Setup OpenInfra Channels","message":"Setup OpenInfra Channels\n\nopenstack-forum, openstack-ptg, openstack-summit and openinfra-summit\nwill now be openinfra-events.\n\nopenstack-diversity will now be openinfra-diversity.\n\nopenstack-foundation will now be openinfra.\n\nopenstack-board will come later after we can get op rights and\nregister the channel.\n\nChange-Id: I4d09979e0c9dac75c7f3d878cde99cdc56b8dbe3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0610c41995ebdea4654703870b6db4617941de41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0610c41995ebdea4654703870b6db4617941de41"}]},"branch":"refs/heads/master"},"f9c31de4c134c880d7ef67c3f074dc4ffcf0a2b8":{"kind":"REWORK","_number":3,"created":"2021-02-09 00:03:04.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/50/774550/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/774550/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/774550/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/774550/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/774550/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/774550/3"}}},"commit":{"parents":[{"commit":"e707d93f2b38b5c1b0645fdc2f9e62951f6ded52","subject":"Merge \"zuul-worker: remove additional install of apt-transport-https\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e707d93f2b38b5c1b0645fdc2f9e62951f6ded52"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-02-08 22:06:02.000000000","tz":-480},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-02-09 00:02:58.000000000","tz":0},"subject":"Setup OpenInfra Channels","message":"Setup OpenInfra Channels\n\nopenstack-forum, openstack-ptg, openstack-summit and openinfra-summit\nwill now be openinfra-events.\n\nopenstack-diversity will now be openinfra-diversity.\n\nopenstack-foundation will now be openinfra.\n\nopenstack-board will come later after we can get op rights and\nregister the channel.\n\nChange-Id: I4d09979e0c9dac75c7f3d878cde99cdc56b8dbe3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f9c31de4c134c880d7ef67c3f074dc4ffcf0a2b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f9c31de4c134c880d7ef67c3f074dc4ffcf0a2b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
