)]}'
{"id":"openstack%2Fproject-config~774906","triplet_id":"openstack%2Fproject-config~master~Ie339113d19fe93bb6f6aff14eab2783df22373a2","project":"openstack/project-config","branch":"master","topic":"test-release-openstack-focal","hashtags":[],"change_id":"Ie339113d19fe93bb6f6aff14eab2783df22373a2","subject":"test-release-openstack: use focal","status":"MERGED","created":"2021-02-10 17:26:03.000000000","updated":"2021-02-10 19:02:12.000000000","submitted":"2021-02-10 19:02:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"774906-test-release-openstack-focal","meta_rev_id":"bc2618a596a2f8538d941b231928f47269f0ef5f","_number":774906,"virtual_id_number":774906,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-10 19:01:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2021-02-10 17:56:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2021-02-10 18:29:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-10 18:51:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-10 18:51:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":30798,"name":"Nathan Weinberg","email":"nweinber@redhat.com","username":"nweinber"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-10 17:28:01.000000000","updated_by":{"_account_id":30798,"name":"Nathan Weinberg","email":"nweinber@redhat.com","username":"nweinber"},"reviewer":{"_account_id":30798,"name":"Nathan Weinberg","email":"nweinber@redhat.com","username":"nweinber"},"state":"CC"},{"updated":"2021-02-10 17:50:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-10 17:56:10.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2021-02-10 18:29:14.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-02-10 18:43:03.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"a8529e5c84c5872d024e9d9ac7723be1ee414dfa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-02-10 17:26:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4aa353030dda44c5f87b50aa567d87f5936946e","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-02-10 17:42:44.000000000","message":"Patch Set 1:\n\nMoreover, all (non deprecated) release-* jobs which deal with artifacts already use ubuntu-focal (inheriting from publish-openstack-artifacts).","accounts_in_message":[],"_revision_number":1},{"id":"7cd37d52807caba97f7cfbbb7688f6019e676241","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 17:50:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/dce586c2efa943108dbb9f375b1678ed : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"b1ac8aab0435dc364514515056846f5786fa0c16","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-02-10 17:56:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b8ccb2144085dcf0e6cc9fae9b04993e3f7c14dc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-10 18:29:14.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nLGTM with a minor suggestion/question, c.f my inline comment.","accounts_in_message":[],"_revision_number":1},{"id":"61bda2b5f3a6bfb0dd9cb8f49b848095f8e07f68","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-02-10 18:31:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"df9e9d63515f4d9ff21294a7dd3c4eb3d1c3fb0e","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-02-10 18:51:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLet\u0027s try this, soon we need to talk about switching the default nodeset in the openstack tenant to ubuntu-focal anyway, so it\u0027s also a good check that it doesn\u0027t break these operations.","accounts_in_message":[],"_revision_number":1},{"id":"7556827663695d711cd31f639c49afdac98e02b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 18:51:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8d70952834063133f0002fd843486298c78412ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 19:01:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ca9d269a8351418389148f1c079cb014 : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":1},{"id":"bc2618a596a2f8538d941b231928f47269f0ef5f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 19:02:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3310a349a4ecf8187f242b9a8fe7d62465f3e083","revisions":{"3310a349a4ecf8187f242b9a8fe7d62465f3e083":{"kind":"REWORK","_number":1,"created":"2021-02-10 17:26:03.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/06/774906/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/06/774906/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/774906/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/774906/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/774906/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/06/774906/1"}}},"commit":{"parents":[{"commit":"23cc8309a7b8f4b3f3771dfbe39363b21c759c40","subject":"Merge \"Add zuul-storage-proxy in zuul namespace\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/23cc8309a7b8f4b3f3771dfbe39363b21c759c40"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2021-02-10 17:23:43.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2021-02-10 17:25:52.000000000","tz":60},"subject":"test-release-openstack: use focal","message":"test-release-openstack: use focal\n\nThe base nodeset is still bionic, but at least cryptography fails\nto build (after https://github.com/pyca/cryptography/issues/5753).\nLong-term is probably the long term fix (the base nodesets\nare going to switch to focal at some point, and the override\ncould be then removed).\n\nChange-Id: Ie339113d19fe93bb6f6aff14eab2783df22373a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3310a349a4ecf8187f242b9a8fe7d62465f3e083"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3310a349a4ecf8187f242b9a8fe7d62465f3e083"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
