)]}'
{"id":"openstack%2Fproject-config~776511","triplet_id":"openstack%2Fproject-config~master~Ic72e324c65ee9d18e9c4cf6627ea6c147b9f484b","project":"openstack/project-config","branch":"master","topic":"change-776511-1","hashtags":[],"change_id":"Ic72e324c65ee9d18e9c4cf6627ea6c147b9f484b","subject":"Remove fedora-31 disk image config","status":"MERGED","created":"2021-02-18 17:30:40.000000000","updated":"2021-03-29 01:29:08.000000000","submitted":"2021-03-29 01:16:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776511-change-776511-1","meta_rev_id":"805b391787d741d3b3a61a14e0de149003fb4183","_number":776511,"virtual_id_number":776511,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-29 01:16:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-29 01:07:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-29 01:07:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-18 17:41:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-19 13:37:12.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2021-02-19 22:00:58.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2021-03-29 00:15:57.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"}],"messages":[{"id":"bb03f4f00fe9bf114fd4506bf6ad6d540bc5a3d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-02-18 17:30:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1e2f49a82ae13d81be216de9d13cb599a43f148","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 17:41:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/163d0c21e63b4b5c9db33757329fcbe2 : SUCCESS in 5m 53s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/2041d3193f754163a0c3af8cb8a4f317 : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":1},{"id":"0e79218d1c272ba1fe9384cce63e9dc192f71f19","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-02-19 13:37:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cf4ce03dcb7970513fcc84f5c4c9ef0e02baaa53","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-02-19 22:00:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d788c1d6b74a8ad9afb5bd2e95917b226020baba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-03-29 00:15:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"47276eea9a6078f749eb11a244b9b997864f0573","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 00:24:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/110159a3c998422aa579d17633a5d04c : SUCCESS in 5m 03s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/516aa9f962a34de7ae86a63609c2a483 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":2},{"id":"245b83a4a1758bfa979272d9996d4458534a7afa","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-03-29 01:07:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nnodes are gone","accounts_in_message":[],"_revision_number":2},{"id":"695c9a7bfea70a430b43286328030b0e2e1f6529","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 01:07:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"33994c8103b24c46d2de51e403596c1fa2c4ceda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 01:16:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1aeb3d23bded46c28a86ec118e387cb5 : SUCCESS in 6m 49s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/6094c3de80c746a2a164d04ebd1b9633 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":2},{"id":"15fd8a9e670e536595aacb0a6fc69cc81c489209","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 01:16:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"805b391787d741d3b3a61a14e0de149003fb4183","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 01:29:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/c0c850cb44b941abb3bc3ee73a067133 : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8c0185da955998775a71d06b708d697435fa005e","revisions":{"dd2810eaf27cd065db4464ef8c3af7616fbcdce8":{"kind":"REWORK","_number":1,"created":"2021-02-18 17:30:40.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/11/776511/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/11/776511/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/776511/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/776511/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/776511/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/11/776511/1"}}},"commit":{"parents":[{"commit":"dcb24931937f6e74d8f32ab42f2c3ce5d5ea62ce","subject":"Stop launch fedora-31 nodes nodepool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dcb24931937f6e74d8f32ab42f2c3ce5d5ea62ce"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2021-02-18 17:30:05.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2021-02-18 17:30:05.000000000","tz":-480},"subject":"Remove fedora-31 disk image config","message":"Remove fedora-31 disk image config\n\nThis is a followup change to the previous change which removed fedora-31\nnode launches. Here we cleanup the disk image configs.\n\nChange-Id: Ic72e324c65ee9d18e9c4cf6627ea6c147b9f484b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dd2810eaf27cd065db4464ef8c3af7616fbcdce8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dd2810eaf27cd065db4464ef8c3af7616fbcdce8"}]},"branch":"refs/heads/master"},"8c0185da955998775a71d06b708d697435fa005e":{"kind":"REWORK","_number":2,"created":"2021-03-29 00:15:57.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/11/776511/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/11/776511/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/776511/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/776511/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/776511/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/11/776511/2"}}},"commit":{"parents":[{"commit":"934b12a7e86555a39d2b652aaef8df470784fcd2","subject":"Stop launch fedora-31 nodes nodepool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/934b12a7e86555a39d2b652aaef8df470784fcd2"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2021-02-18 17:30:05.000000000","tz":-480},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-03-29 00:15:29.000000000","tz":660},"subject":"Remove fedora-31 disk image config","message":"Remove fedora-31 disk image config\n\nThis is a followup change to the previous change which removed fedora-31\nnode launches. Here we cleanup the disk image configs.\n\nChange-Id: Ic72e324c65ee9d18e9c4cf6627ea6c147b9f484b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8c0185da955998775a71d06b708d697435fa005e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8c0185da955998775a71d06b708d697435fa005e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
