)]}'
{"id":"openstack%2Fproject-config~776644","triplet_id":"openstack%2Fproject-config~master~I789182053c7614e0e1197865642e463bd4f3d412","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I789182053c7614e0e1197865642e463bd4f3d412","subject":"Deprecate monasca-ceilometer","status":"MERGED","created":"2021-02-19 14:03:55.000000000","updated":"2021-02-19 19:40:10.000000000","submitted":"2021-02-19 19:12:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776644","meta_rev_id":"1ed60b4048f98bff9cc036777df3534fe1d8a081","_number":776644,"virtual_id_number":776644,"owner":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-19 19:12:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":1,"date":"2021-02-19 18:45:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-19 18:59:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-19 18:59:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-19 14:08:04.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2021-02-19 14:08:04.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-02-19 14:18:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-19 18:59:48.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"87d540a74458a369d679b851a91ffb67e1f9c61a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-02-19 14:03:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7859efa990e686ba1b2cca823c554b581dc211f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-02-19 14:05:09.000000000","message":"Topic ceilometer removed","accounts_in_message":[],"_revision_number":1},{"id":"e8bfe8d94a79532c43661df9c7fbfe6fae91041e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 14:18:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/00e3072038d642a4beb2edba6e693e5b : SUCCESS in 6m 19s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/1eb85cce555f4c7f999620a6065ba30b : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":1},{"id":"697150f69feaf863da4793b4e2b82ad461c768d6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-02-19 15:52:37.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cf1a80b48f017d8ab73fa47f56753dd2249b284e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-02-19 16:03:23.000000000","message":"Patch Set 1: Code-Review-1\n\nyou can remove this also as no release is going to happen now. keeping official-openstack-repo-jobs is only needed","accounts_in_message":[],"_revision_number":1},{"id":"34e7b64c3e55a84baffdda7eb2ca7b746033132e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-02-19 16:06:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"deb3bbaed34443eb1921928a6af52a91819c6621","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-02-19 16:09:51.000000000","message":"Topic ceilometer removed","accounts_in_message":[],"_revision_number":2},{"id":"1ca09c6214fa1b3598da142b5c9818136b5982f4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-02-19 16:12:47.000000000","message":"Patch Set 2: Code-Review+1\n\nthanks","accounts_in_message":[],"_revision_number":2},{"id":"7b4b4f8dfe79745ba5a231645dd3a21f6bfa1c66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 16:19:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/af517dfb2e864c03a49a1c44609f069d : POST_FAILURE in 5m 44s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/ad9c34ad5325427ab86d7d1b67e26793 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":2},{"id":"38b6e5d9993ae5db3eb63425f13f66b3e6a2fa4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-02-19 16:31:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a5e3352d9a5c7f9b7bb2eeefdab8c22ff627b4a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 16:31:39.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  expected a list for dictionary value @ data[\u0027templates\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      name: openstack/monasca-ceilometer\n      templates:\n        official-openstack-repo-jobs\n      check:\n        jobs:\n          - noop:\n              branches: master\n      gate:\n        jobs:\n          - noop:\n              branches: master\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 3083, column 3","accounts_in_message":[],"_revision_number":3},{"id":"57fa89768018ee3dd03cd41bca1ece11ecc19dde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-02-19 16:45:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"696bf8853aa7d94deaf612fb7f7eaf327d9b72a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 17:14:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/06492d6cdc4f4f079fccc164a2fc4cb6 : POST_FAILURE in 7m 46s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/c65a52819e1841bc87b670473bbf8eae : POST_FAILURE in 5m 01s","accounts_in_message":[],"_revision_number":4},{"id":"f78e56c79d950bbd04d171b26c65f91513938a62","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-02-19 17:45:25.000000000","message":"Patch Set 4:\n\nits all good here, job failing due to infra issue.","accounts_in_message":[],"_revision_number":4},{"id":"b97761043b97aabd147961a7d25c58ba49b21358","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-02-19 18:01:00.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"2e39f19c429cba306ea82abc3cdc2b42cbe151d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 18:20:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/dcf6608f198642b5989d16c77d12bfff : SUCCESS in 6m 06s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/0066ecf390164635bfe2b4de902e90ae : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":4},{"id":"cbb3164125225a92f0d66e7c76a241687c976f09","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-02-19 18:45:13.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"29d75e3aaaa5ccaba360b69f760466aa0d3da88f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-02-19 18:59:48.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"419afbb3e5441a741a81651e16b7ad8757c2cdd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 19:00:11.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d834f6f614b239dc7c63926adf48cbdb533d8164","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 19:12:25.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a4fc2f33674b413c819d73d608174724 : SUCCESS in 7m 26s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/408fc7346ee940c79cb8190c9570da1e : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":4},{"id":"e575b4c3ad3f040f721174239571189f8a542840","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 19:12:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"1ed60b4048f98bff9cc036777df3534fe1d8a081","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 19:40:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/70f0b82b153a4f7684b68b946308553e : SUCCESS in 15m 53s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b6d8c24b6ae7d4591325df9973f9888ec47c657d","revisions":{"5724ca0336eac1ffb5df699b5f2b2cdaaa158024":{"kind":"REWORK","_number":1,"created":"2021-02-19 14:03:55.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/44/776644/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/44/776644/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/776644/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/776644/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/776644/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/44/776644/1"}}},"commit":{"parents":[{"commit":"795751324562263d6dddb1860dd64e0c1f86cc55","subject":"Merge \"Add ansible-role-pki to zuul\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/795751324562263d6dddb1860dd64e0c1f86cc55"}]}],"author":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2021-02-19 14:03:07.000000000","tz":60},"committer":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2021-02-19 14:03:07.000000000","tz":60},"subject":"Deprecate monasca-ceilometer","message":"Deprecate monasca-ceilometer\n\nChange-Id: I789182053c7614e0e1197865642e463bd4f3d412\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5724ca0336eac1ffb5df699b5f2b2cdaaa158024"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5724ca0336eac1ffb5df699b5f2b2cdaaa158024"}]},"branch":"refs/heads/master"},"f93b805ca941da0216655dcae6f0f3d6520a01db":{"kind":"REWORK","_number":2,"created":"2021-02-19 16:06:55.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/44/776644/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/44/776644/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/776644/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/776644/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/776644/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/44/776644/2"}}},"commit":{"parents":[{"commit":"795751324562263d6dddb1860dd64e0c1f86cc55","subject":"Merge \"Add ansible-role-pki to zuul\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/795751324562263d6dddb1860dd64e0c1f86cc55"}]}],"author":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2021-02-19 14:03:07.000000000","tz":60},"committer":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2021-02-19 16:06:25.000000000","tz":60},"subject":"Deprecate monasca-ceilometer","message":"Deprecate monasca-ceilometer\n\nChange-Id: I789182053c7614e0e1197865642e463bd4f3d412\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f93b805ca941da0216655dcae6f0f3d6520a01db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f93b805ca941da0216655dcae6f0f3d6520a01db"}]},"branch":"refs/heads/master"},"1a2c6d333ad6820ca3bb0677caf418cbc4901865":{"kind":"REWORK","_number":3,"created":"2021-02-19 16:31:15.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/44/776644/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/44/776644/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/776644/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/776644/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/776644/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/44/776644/3"}}},"commit":{"parents":[{"commit":"795751324562263d6dddb1860dd64e0c1f86cc55","subject":"Merge \"Add ansible-role-pki to zuul\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/795751324562263d6dddb1860dd64e0c1f86cc55"}]}],"author":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2021-02-19 14:03:07.000000000","tz":60},"committer":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2021-02-19 16:30:52.000000000","tz":60},"subject":"Deprecate monasca-ceilometer","message":"Deprecate monasca-ceilometer\n\nChange-Id: I789182053c7614e0e1197865642e463bd4f3d412\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1a2c6d333ad6820ca3bb0677caf418cbc4901865"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1a2c6d333ad6820ca3bb0677caf418cbc4901865"}]},"branch":"refs/heads/master"},"b6d8c24b6ae7d4591325df9973f9888ec47c657d":{"kind":"REWORK","_number":4,"created":"2021-02-19 16:45:16.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/44/776644/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/44/776644/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/776644/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/776644/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/776644/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/44/776644/4"}}},"commit":{"parents":[{"commit":"795751324562263d6dddb1860dd64e0c1f86cc55","subject":"Merge \"Add ansible-role-pki to zuul\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/795751324562263d6dddb1860dd64e0c1f86cc55"}]}],"author":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2021-02-19 14:03:07.000000000","tz":60},"committer":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2021-02-19 16:44:44.000000000","tz":60},"subject":"Deprecate monasca-ceilometer","message":"Deprecate monasca-ceilometer\n\nChange-Id: I789182053c7614e0e1197865642e463bd4f3d412\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b6d8c24b6ae7d4591325df9973f9888ec47c657d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b6d8c24b6ae7d4591325df9973f9888ec47c657d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
