)]}'
{"id":"openstack%2Fproject-config~777480","triplet_id":"openstack%2Fproject-config~master~Ib631e030955bcbd0fcdac5492bf6655dff6b131b","project":"openstack/project-config","branch":"master","topic":"gerrit-zuul-branch","hashtags":[],"change_id":"Ib631e030955bcbd0fcdac5492bf6655dff6b131b","subject":"zuul-results-summary: set default branch","status":"MERGED","created":"2021-02-24 23:07:59.000000000","updated":"2021-02-25 01:27:33.000000000","submitted":"2021-02-25 01:27:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777480-gerrit-zuul-branch","meta_rev_id":"c3fd85258413fd91e806948cae4f5e870730a600","_number":777480,"virtual_id_number":777480,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-25 01:27:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-25 01:16:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-25 01:16:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-24 23:09:04.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2021-02-24 23:16:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"227469d79f9b48dd2a7017965b40ba5bbbf49cc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-02-24 23:07:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"676634c8c9948b43829908ad8f8910c65afade1c","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2021-02-24 23:09:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fc8a79cb1ef948abcc749f38b7235d46aa8690dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 23:16:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b7ab4865d8a6466fadb995bc574234fa : FAILURE in 4m 41s","accounts_in_message":[],"_revision_number":1},{"id":"da9d0e345f1d6d7179bc74acbc738f36874bebf7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-02-24 23:23:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d6730923e8d5acb083eb8b4f98ff89b4d5ede76f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 23:32:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b0e73f3251044fc7b9a726f6862f8be6 : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":2},{"id":"00a36b268474b998a7873a99b6a068076479e7c9","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-02-25 01:05:14.000000000","message":"Patch Set 2:\n\nIt seems I04f5be25414c923dcd25cc490b4b0bff5a48eefa is not sufficient","accounts_in_message":[],"_revision_number":2},{"id":"84ff49f767b25b30397c6d5e0ad70ec7bae1fd68","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-02-25 01:16:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nthe conclusion seems to be that we require this too.  since the last review it just moved underneath ansible to keep ordering","accounts_in_message":[],"_revision_number":2},{"id":"12ac981bf80aed297b4800b2817c8fe6d484803a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 01:16:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a7e7a1a6211420949d18954f3c2ca688f7ed4254","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 01:27:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6a78784d53bf4b30a39430f10a86c4f7 : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":2},{"id":"c3fd85258413fd91e806948cae4f5e870730a600","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 01:27:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2dccaa84325f53d447dea68dba04f07e06a3e752","revisions":{"dde1adb0a089587c4105e1ecddb1ee3a284d20d2":{"kind":"REWORK","_number":1,"created":"2021-02-24 23:07:59.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/80/777480/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/80/777480/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/777480/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/777480/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/777480/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/80/777480/1"}}},"commit":{"parents":[{"commit":"d93652b5e30543c4a8e4b92cefd6a1e5d15f1bf6","subject":"Merge \"Deprecate monasca-log-api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d93652b5e30543c4a8e4b92cefd6a1e5d15f1bf6"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-02-24 23:07:13.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-02-24 23:07:13.000000000","tz":660},"subject":"zuul-results-summary: set default branch","message":"zuul-results-summary: set default branch\n\nThis project has a \"main\" branch, not master.  Set the default branch.\n\nChange-Id: Ib631e030955bcbd0fcdac5492bf6655dff6b131b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dde1adb0a089587c4105e1ecddb1ee3a284d20d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dde1adb0a089587c4105e1ecddb1ee3a284d20d2"}]},"branch":"refs/heads/master"},"2dccaa84325f53d447dea68dba04f07e06a3e752":{"kind":"REWORK","_number":2,"created":"2021-02-24 23:23:06.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/80/777480/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/80/777480/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/777480/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/777480/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/777480/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/80/777480/2"}}},"commit":{"parents":[{"commit":"d93652b5e30543c4a8e4b92cefd6a1e5d15f1bf6","subject":"Merge \"Deprecate monasca-log-api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d93652b5e30543c4a8e4b92cefd6a1e5d15f1bf6"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-02-24 23:07:13.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-02-24 23:22:52.000000000","tz":660},"subject":"zuul-results-summary: set default branch","message":"zuul-results-summary: set default branch\n\nThis project has a \"main\" branch, not master.  Set the default branch.\n\nChange-Id: Ib631e030955bcbd0fcdac5492bf6655dff6b131b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2dccaa84325f53d447dea68dba04f07e06a3e752"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2dccaa84325f53d447dea68dba04f07e06a3e752"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
