)]}'
{"id":"openstack%2Fproject-config~777675","triplet_id":"openstack%2Fproject-config~master~I6a1348396f5a374bf2dc3cd7ac2f9ffe25e5b0b2","project":"openstack/project-config","branch":"master","topic":"change-777675-1","hashtags":[],"change_id":"I6a1348396f5a374bf2dc3cd7ac2f9ffe25e5b0b2","subject":"Retire ara projects","status":"MERGED","created":"2021-02-25 20:35:55.000000000","updated":"2021-07-26 18:05:34.000000000","submitted":"2021-07-26 17:05:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777675-change-777675-1","meta_rev_id":"817e544d6e7f6842d732fa3e541628e0fae1f443","_number":777675,"virtual_id_number":777675,"owner":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-26 17:05:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2021-07-26 10:45:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-26 16:58:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-26 16:58:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-25 20:47:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-01 00:26:11.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"CC"},{"updated":"2021-07-23 01:10:34.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2021-07-23 06:49:48.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2021-07-26 16:58:05.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"4e9949f7f1c76fc956353b58cf9eddf430cd6d4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"date":"2021-02-25 20:35:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7bfda59b01f0a01586190312e2849e3a4d252cf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 20:47:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0b08cca3433f456a890c1f85df0e7ab6 : SUCCESS in 5m 25s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/4a8817af6bb04ae298d8382c171e601e : SUCCESS in 5m 08s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/3101e5f48a714b24a3f3129f9514ebad : SUCCESS in 7m 59s","accounts_in_message":[],"_revision_number":1},{"id":"a76445d51574cc2794c207ba07fcda1972c4f16a","author":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"date":"2021-02-25 21:51:15.000000000","message":"Patch Set 1: Workflow-1\n\nI found a recommendation to check in codesearch in the retiring docs [0] and found a few matches:\n1) https://opendev.org/opendev/system-config/src/branch/master/zuul.d/system-config-run.yaml#L83-L86\n2) https://opendev.org/openstack/openstack-ansible/src/branch/master/scripts/scripts-library.sh#L202\n3) https://opendev.org/openstack/openstack-ansible-tests/src/branch/master/test-ansible-env-prep.sh#L199\n4) https://opendev.org/zuul/zuul/src/branch/master/playbooks/zuul-stream/pre.yaml#L10-L11\n\n[0]: https://docs.openstack.org/project-team-guide/repository.html#retiring-a-repository","accounts_in_message":[],"_revision_number":1},{"id":"a6e771173acc07bf158f4657d7a8f6930063b861","author":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"date":"2021-02-25 22:40:45.000000000","message":"Patch Set 1:\n\n\u003e 1) https://opendev.org/opendev/system-config/src/branch/master/zuul.d/system-config-run.yaml#L83-L86\n\nThe required-projects adds ara to src/opendev.org/recordsansible/ara which is used later: https://opendev.org/opendev/system-config/src/branch/master/zuul.d/system-config-run.yaml#L100\n\nI found out that I forgot to retire the stable/0.x branch so this still works for now.\nIt\u0027s available on github: https://github.com/ansible-community/ara/tree/stable/0.x but I don\u0027t maintain it anymore.\n\nTo keep the current approach would mean to add ara\u0027s github repo in the Zuul included projects (like ansible/ansible is) and then update the required-project with the new name -- or install from source by cloning the repository from github first (or install directly from pip with git+https://...?)\n\nDo you have a preference ?\n\n\u003e 2) https://opendev.org/openstack/openstack-ansible/src/branch/master/scripts/scripts-library.sh#L202\n\u003e 3) https://opendev.org/openstack/openstack-ansible-tests/src/branch/master/test-ansible-env-prep.sh#L199\n\nThese weren\u0027t required-projects but scripts that provided support for leveraging Depends-On based on the path opendev.org/recordsansible/ara. I don\u0027t think we\u0027ve used it that much. I\u0027ve proposed patches to simplify the installation:\n- https://review.opendev.org/c/openstack/openstack-ansible/+/777696\n- https://review.opendev.org/c/openstack/openstack-ansible-tests/+/777697\n\n\u003e 4) https://opendev.org/zuul/zuul/src/branch/master/playbooks/zuul-stream/pre.yaml#L10-L11\n\nThis is used by zuul-stream-functional which is the parent to zuul-stream-functional-2.8 and 2.9.\nIt uses bindep to install these packages: https://github.com/ansible-community/ara/blob/stable/0.x/bindep.txt\n\nTo keep the current approach we could add a task that clones the repository before the bindep role is called and set the right path.\n\nWould that work ?","accounts_in_message":[],"_revision_number":1},{"id":"3ef20b31c638f8b6987085a2ab81c133a8c993fc","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-06-01 00:26:11.000000000","message":"Patch Set 1:\n\n\u003e \u003e 4) https://opendev.org/zuul/zuul/src/branch/master/playbooks/zuul-stream/pre.yaml#L10-L11\n\u003e \n\u003e This is used by zuul-stream-functional which is the parent to zuul-stream-functional-2.8 and 2.9.\n\u003e It uses bindep to install these packages: https://github.com/ansible-community/ara/blob/stable/0.x/bindep.txt\n\nThe actual dependency seems to have been dropped with https://review.opendev.org/c/zuul/zuul/+/631930 \n\nI\u0027ve proposed https://review.opendev.org/c/zuul/zuul/+/793868 to clean this up\n\nAlthough the zuul managed ansible does install a capped version; that\u0027s another problem though 😊","accounts_in_message":[],"_revision_number":1},{"id":"f789ec6731f573ce5e5a9d614e8709837a4c700d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-07-23 01:10:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"876366dbc2694c6486597aec2aad7637aa1964af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 01:18:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d48e2e390aec438e9a9d5231cd8ef981 : SUCCESS in 6m 21s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/a0b6dfc84a4e40a59f1784f68abd21bd : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":2},{"id":"5b247fa73713b2f709d5718f4d6712d991bce22b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-07-23 06:49:48.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ee6dc9f0dc099aad0bb2a2545c92095c0ced0bc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-07-26 10:44:21.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"540a917941efa828f4737a9bdeff3a7d621d93ee","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-07-26 10:44:46.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nFixed the missing acl","accounts_in_message":[],"_revision_number":3},{"id":"9abb29c78232cfbef08c1127d82a07e4a64eb406","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-07-26 10:45:20.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"083981b17266ce03183cc8c0329be7288c4860ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 10:58:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f0f63b214c034a50b73431582c0ef4ab : SUCCESS in 7m 29s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/3993df963097406aabf33315ddb62ea4 : SUCCESS in 7m 09s","accounts_in_message":[],"_revision_number":3},{"id":"c291f306cd1583ef4c2df6b6bdd3a3ae6ff611f4","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-07-26 16:58:05.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"40aa57c9d442cf333da190a27700d2633e9282fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 16:58:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"da8569b19d1a2087c7c077cda9177263a2a7f655","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 17:05:44.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b375a36c3798421bba4c2bb01e2477e8 : SUCCESS in 6m 18s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/ff7cc1dda51f4818b312bfb21a25b808 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":3},{"id":"34a8a7fa39ec8841fe3c37de971e36a9d702305c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 17:05:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"0d1b1e1567a0ef35c96179a1db64be2773a90475","author":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"date":"2021-07-26 17:11:56.000000000","message":"Patch Set 3:\n\nThanks Andreas, Ian and Jeremy for wrapping this up !","accounts_in_message":[],"_revision_number":3},{"id":"817e544d6e7f6842d732fa3e541628e0fae1f443","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 18:05:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/0685015ebdda4e00864b5306993bc81f : SUCCESS in 3m 17s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/09be72955c424c53a5b2c485e08fcb33 : SUCCESS in 7m 13s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"506e1a97e4562b7fb1f9a42505c8f5469688912e","revisions":{"4d5a51d6da4b67656e582c5fc979969c44d560fc":{"kind":"REWORK","_number":1,"created":"2021-02-25 20:35:55.000000000","uploader":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"ref":"refs/changes/75/777675/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/75/777675/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/777675/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/777675/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/777675/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/75/777675/1"}}},"commit":{"parents":[{"commit":"e2ddf3fc7c15403dc23e2d96f3bb7bc49d4be0e0","subject":"Merge \"Create inspur/venus-dashboard repository\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e2ddf3fc7c15403dc23e2d96f3bb7bc49d4be0e0"}]}],"author":{"name":"David Moreau Simard","email":"moi@dmsimard.com","date":"2021-02-25 20:33:12.000000000","tz":-300},"committer":{"name":"David Moreau Simard","email":"moi@dmsimard.com","date":"2021-02-25 20:33:12.000000000","tz":-300},"subject":"Retire ara projects","message":"Retire ara projects\n\nThank you for hosting ara for many years.\n\nCode reviews for these projects are moving to GitHub [1].\n\n[1]: http://lists.openstack.org/pipermail/openstack-discuss/2021-February/020705.html\n\nChange-Id: I6a1348396f5a374bf2dc3cd7ac2f9ffe25e5b0b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4d5a51d6da4b67656e582c5fc979969c44d560fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4d5a51d6da4b67656e582c5fc979969c44d560fc"}]},"branch":"refs/heads/master"},"d16f90577a7e863f29d522432e8ca733d85323aa":{"kind":"REWORK","_number":2,"created":"2021-07-23 01:10:34.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/75/777675/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/75/777675/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/777675/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/777675/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/777675/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/75/777675/2"}}},"commit":{"parents":[{"commit":"1744e6e3d655327b7e1ead975f1ad17e2bb4fbdb","subject":"Merge \"Retire django-openstack-auth\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1744e6e3d655327b7e1ead975f1ad17e2bb4fbdb"}]}],"author":{"name":"David Moreau Simard","email":"moi@dmsimard.com","date":"2021-02-25 20:33:12.000000000","tz":-300},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-07-23 01:10:07.000000000","tz":600},"subject":"Retire ara projects","message":"Retire ara projects\n\nThank you for hosting ara for many years.\n\nCode reviews for these projects are moving to GitHub [1].\n\n[1]: http://lists.openstack.org/pipermail/openstack-discuss/2021-February/020705.html\n\nChange-Id: I6a1348396f5a374bf2dc3cd7ac2f9ffe25e5b0b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d16f90577a7e863f29d522432e8ca733d85323aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d16f90577a7e863f29d522432e8ca733d85323aa"}]},"branch":"refs/heads/master"},"506e1a97e4562b7fb1f9a42505c8f5469688912e":{"kind":"REWORK","_number":3,"created":"2021-07-26 10:44:21.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/75/777675/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/75/777675/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/777675/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/777675/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/777675/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/75/777675/3"}}},"commit":{"parents":[{"commit":"1744e6e3d655327b7e1ead975f1ad17e2bb4fbdb","subject":"Merge \"Retire django-openstack-auth\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1744e6e3d655327b7e1ead975f1ad17e2bb4fbdb"}]}],"author":{"name":"David Moreau Simard","email":"moi@dmsimard.com","date":"2021-02-25 20:33:12.000000000","tz":-300},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2021-07-26 10:44:14.000000000","tz":0},"subject":"Retire ara projects","message":"Retire ara projects\n\nThank you for hosting ara for many years.\n\nCode reviews for these projects are moving to GitHub [1].\n\n[1]: http://lists.openstack.org/pipermail/openstack-discuss/2021-February/020705.html\n\nChange-Id: I6a1348396f5a374bf2dc3cd7ac2f9ffe25e5b0b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/506e1a97e4562b7fb1f9a42505c8f5469688912e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/506e1a97e4562b7fb1f9a42505c8f5469688912e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
