)]}'
{"id":"openstack%2Fproject-config~778590","triplet_id":"openstack%2Fproject-config~master~I67912064487598c0e5b4ce3001276f42e0ebcad1","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I67912064487598c0e5b4ce3001276f42e0ebcad1","subject":"Add custom cirros image with ahci module enabled to cache","status":"MERGED","created":"2021-03-03 22:28:19.000000000","updated":"2021-03-08 08:43:53.000000000","submitted":"2021-03-04 01:00:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778590","meta_rev_id":"70be0ea024b2681a95ab73fe08b5280b5d9612bf","_number":778590,"virtual_id_number":778590,"owner":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-04 01:00:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2021-03-04 00:49:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-03 23:02:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2021-03-04 00:49:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-03 22:40:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-03 23:02:19.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2021-03-04 00:49:47.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"aa602be814ddf2aa819910da9f4921bbf5eff246","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-03-03 22:28:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54ef3fcf4ed6049c758b77e0302d1210698df2aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 22:40:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f81abc9f360e4326905a51dc87b4eae2 : SUCCESS in 5m 57s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/e6ae294418d94f3980d6989832e33550 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":1},{"id":"5808e2a930e7a20a51b1e192611aae9b61a42263","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-03-03 23:02:19.000000000","message":"Patch Set 1: Code-Review+2\n\nThis will only increase our raw image size by 16MiB, so I\u0027m not overly concerned.","accounts_in_message":[],"_revision_number":1},{"id":"0f6a364fe64c1ec139f133a2ccff6d2b4bbd9e19","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-03-04 00:49:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nAre we able to replace all cirros use with this image too? I\u0027m approving it to keep things moving, but thinking out loud maybe we can use this as the cirros image.","accounts_in_message":[],"_revision_number":1},{"id":"c61225200d07d04c894524f6d6fe432043eef1a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 00:50:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5392f06ad190e02c5945fb322f5cd651bdf48d43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 01:00:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b2d8388db9ee4e9796c6c0488cc98266 : SUCCESS in 5m 51s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/624a43397ce14e279ad241bff4af4842 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":1},{"id":"c2a56587663e02a6b17ddbbc9e83b620db91f4d8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 01:00:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e2d004df6e0e7bf2e44c29fa314037a2774bd4f7","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 01:17:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/257b4614ea2046759c726ec3025e5a42 : SUCCESS in 7m 29s","accounts_in_message":[],"_revision_number":1},{"id":"2f754d64bbc7d91f45338533fff1a7ba77390e43","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-03-04 10:55:22.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review+2 Workflow+1\n\u003e \n\u003e Are we able to replace all cirros use with this image too? I\u0027m approving it to keep things moving, but thinking out loud maybe we can use this as the cirros image.\n\nAs it isn\u0027t formally released yet by the Cirros folks and we are so close to M3 I wanted to limit the impact of any issues this cases to just the nova-next job. I can\u0027t think that it will cause any issues however as it\u0027s just adding the ahci kernel module, I\u0027m just being super cautious given where we are in the cycle.","accounts_in_message":[],"_revision_number":1},{"id":"70be0ea024b2681a95ab73fe08b5280b5d9612bf","tag":"autogenerated:gerrit:revert","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-03-08 08:43:53.000000000","message":"Created a revert of this change as I5332d0e47ad863ca9795a8b0b86b73156621622d","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5ee0780486be4340c3585a115018dd4b192d5b72","revisions":{"5ee0780486be4340c3585a115018dd4b192d5b72":{"kind":"REWORK","_number":1,"created":"2021-03-03 22:28:19.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/90/778590/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/90/778590/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/778590/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/778590/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/778590/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/90/778590/1"}}},"commit":{"parents":[{"commit":"52e95f859ccb7c941182b94e61b1ce7819dc1024","subject":"Add tripleo-ci-health-queries to zuul","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/52e95f859ccb7c941182b94e61b1ce7819dc1024"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-03-03 22:23:54.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-03-03 22:28:05.000000000","tz":0},"subject":"Add custom cirros image with ahci module enabled to cache","message":"Add custom cirros image with ahci module enabled to cache\n\nAs discussed on the ML [1] the nova-next job is looking to start testing\nthe q35 machine type. In order to do this *before* the next Cirros\nrelease a custom dev build of the Cirros image has been built with the\nahci module included, as is now required to allow for SATA based config\ndrives to work.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2021-March/020823.html\n\nChange-Id: I67912064487598c0e5b4ce3001276f42e0ebcad1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5ee0780486be4340c3585a115018dd4b192d5b72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5ee0780486be4340c3585a115018dd4b192d5b72"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
