)]}'
{"id":"openstack%2Fproject-config~780211","triplet_id":"openstack%2Fproject-config~master~I9faa96360bdc3eebb12f88f5f53d0ae13e0e398b","project":"openstack/project-config","branch":"master","topic":"new-project","hashtags":[],"change_id":"I9faa96360bdc3eebb12f88f5f53d0ae13e0e398b","subject":"Add Magnum charms to OpenStack charms","status":"MERGED","created":"2021-03-12 09:50:53.000000000","updated":"2021-03-15 15:11:19.000000000","submitted":"2021-03-15 14:20:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780211-new-project","meta_rev_id":"46d5f2dd5005d35d27ac15e73022bde5d7fb66bc","_number":780211,"virtual_id_number":780211,"owner":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-15 14:19:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-15 14:07:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-15 14:07:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-12 10:01:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-15 14:07:11.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"a62792ff8dae1c0fea3cb9dca2d4ff60efb42359","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2021-03-12 09:50:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c40305fae491be1c867bb66c35493467b966006b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2021-03-12 09:54:55.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a1693babb7216f0b69eb82dc6dd40aed0898cfc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 10:01:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d412ff69b8e3457b861fbb18dec4c79b : SUCCESS in 5m 30s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/1eeb6a98cfd148e6abcd029c2eb71218 : FAILURE in 5m 21s","accounts_in_message":[],"_revision_number":2},{"id":"9a05c2d404f2d96b7c7926bf97754bc62f2592b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2021-03-15 10:17:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b2e68bed78638554daef200d16fd4f28c72a0b17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 10:25:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f73a97e46f7e4acf874689161435d243 : SUCCESS in 5m 18s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/40f10df710374bb7b2827670daed8904 : FAILURE in 4m 16s","accounts_in_message":[],"_revision_number":3},{"id":"f88a905a3bc9d408f7b44b4b751bf910ff80b7d4","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2021-03-15 10:42:09.000000000","message":"Patch Set 3:\n\n.zuul.yaml has now been removed from charm-magnum.git. Re-triggering the CI","accounts_in_message":[],"_revision_number":3},{"id":"21184f85cf244d8ed1019fa3364edc2ca6c35722","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2021-03-15 10:42:26.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"0572250c20284c1b0db33c31267586bc67df0f4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 10:54:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7d6b3fb669e9440eb33e7b54567e7d60 : SUCCESS in 5m 41s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f028c3e02d114f449a658b354290c3b9 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":3},{"id":"ff9657487212726d98c3582f53e9c65cbf9f08a3","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-03-15 14:07:11.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"38770776ef89e05b1102479124914f2018f8de41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 14:07:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"19fe47c5a7f880abaefca18ad0caf9521c016ac3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 14:19:54.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/aaac78020fbf48d08382915cce61c5dc : SUCCESS in 6m 52s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/1eb08d9e7385412b8d94f5e546a7b194 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":3},{"id":"c4cc8b528fa5602326b7d96b748803862343f424","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 14:20:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"46d5f2dd5005d35d27ac15e73022bde5d7fb66bc","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 15:11:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/c62b7e75a27246538054e985d6d45c9b : SUCCESS in 24m 07s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/edd3121c7780420ea9bdfd54348e8314 : SUCCESS in 12m 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"04cefa3f77fbbd28cb7a3b703c397c80d0391565","revisions":{"1c75897ed62abd7199503998bdfc4be9fdcc050f":{"kind":"REWORK","_number":1,"created":"2021-03-12 09:50:53.000000000","uploader":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"ref":"refs/changes/11/780211/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/11/780211/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/780211/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/780211/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/780211/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/11/780211/1"}}},"commit":{"parents":[{"commit":"9833c66cb827c2acc4f293ee3d7b818808543272","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9833c66cb827c2acc4f293ee3d7b818808543272"}]}],"author":{"name":"Aurelien Lourot","email":"aurelien.lourot@canonical.com","date":"2021-03-12 09:50:41.000000000","tz":60},"committer":{"name":"Aurelien Lourot","email":"aurelien.lourot@canonical.com","date":"2021-03-12 09:50:41.000000000","tz":60},"subject":"Add Magnum charm to OpenStack charms","message":"Add Magnum charm to OpenStack charms\n\nAdd new charm that provides the OpenStack Magnum service.\n\nChange-Id: I9faa96360bdc3eebb12f88f5f53d0ae13e0e398b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1c75897ed62abd7199503998bdfc4be9fdcc050f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1c75897ed62abd7199503998bdfc4be9fdcc050f"}]},"branch":"refs/heads/master"},"2d4d792090c70e7dfc0951703ad373a2c13288ab":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-03-12 09:54:55.000000000","uploader":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"ref":"refs/changes/11/780211/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/11/780211/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/780211/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/780211/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/780211/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/11/780211/2"}}},"commit":{"parents":[{"commit":"9833c66cb827c2acc4f293ee3d7b818808543272","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9833c66cb827c2acc4f293ee3d7b818808543272"}]}],"author":{"name":"Aurelien Lourot","email":"aurelien.lourot@canonical.com","date":"2021-03-12 09:50:41.000000000","tz":60},"committer":{"name":"Aurelien Lourot","email":"aurelien.lourot@canonical.com","date":"2021-03-12 09:54:43.000000000","tz":60},"subject":"Add Magnum charm to OpenStack charms","message":"Add Magnum charm to OpenStack charms\n\nAdd new charm that provides the OpenStack Magnum service.\n\nChange-Id: I9faa96360bdc3eebb12f88f5f53d0ae13e0e398b\nRequired-By: I0649a57fb69510f0f6223854f2066411409c3e56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2d4d792090c70e7dfc0951703ad373a2c13288ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2d4d792090c70e7dfc0951703ad373a2c13288ab"}]},"branch":"refs/heads/master"},"04cefa3f77fbbd28cb7a3b703c397c80d0391565":{"kind":"REWORK","_number":3,"created":"2021-03-15 10:17:37.000000000","uploader":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"ref":"refs/changes/11/780211/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/11/780211/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/780211/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/780211/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/780211/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/11/780211/3"}}},"commit":{"parents":[{"commit":"9833c66cb827c2acc4f293ee3d7b818808543272","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9833c66cb827c2acc4f293ee3d7b818808543272"}]}],"author":{"name":"Aurelien Lourot","email":"aurelien.lourot@canonical.com","date":"2021-03-12 09:50:41.000000000","tz":60},"committer":{"name":"Aurelien Lourot","email":"aurelien.lourot@canonical.com","date":"2021-03-15 10:17:25.000000000","tz":60},"subject":"Add Magnum charms to OpenStack charms","message":"Add Magnum charms to OpenStack charms\n\nAdd new charms that provides the OpenStack Magnum service\nand its dashboard.\n\nChange-Id: I9faa96360bdc3eebb12f88f5f53d0ae13e0e398b\nRequired-By: I0649a57fb69510f0f6223854f2066411409c3e56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/04cefa3f77fbbd28cb7a3b703c397c80d0391565"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/04cefa3f77fbbd28cb7a3b703c397c80d0391565"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
