)]}'
{"id":"openstack%2Fproject-config~780984","triplet_id":"openstack%2Fproject-config~master~I9792620eea81a07b6cbbfee37c08807114d2b390","project":"openstack/project-config","branch":"master","topic":"nl02-04.opendev.org-configs","hashtags":[],"change_id":"I9792620eea81a07b6cbbfee37c08807114d2b390","subject":"Remove unused nl0X.openstack.org config files","status":"MERGED","created":"2021-03-16 21:56:58.000000000","updated":"2021-03-18 21:05:40.000000000","submitted":"2021-03-18 20:53:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780984-nl02-04.opendev.org-configs","meta_rev_id":"6378298a681dea4beea5b54aa0c05db9ce612bc2","_number":780984,"virtual_id_number":780984,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-18 20:53:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-18 20:39:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-18 20:42:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-16 22:03:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-18 20:39:28.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"}],"messages":[{"id":"c0287507976817748acbf6614d0b168aa65c6b1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-03-16 21:56:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ecee5dcb251727b8d06dd48180c68dbf7b7ed39c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 22:03:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3ea7236e18db4bd18d28cd007676d900 : SUCCESS in 5m 53s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/10841c7b9ff445f0a5ecd23582bb1dd1 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"e6a7c56691898ae44d6c9e4f63cbdbe2662e14fe","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-03-17 22:50:47.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"9b7d6a3a07508364a1ffb974d2aced58bd17f6e8","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-03-18 17:16:58.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"23c4b152e5e8f6258df52166b6a676e9d60138fa","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-03-18 20:39:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dc0b26e0cd2742e054e051848f0af476c6abd460","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-03-18 20:42:54.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"568c347b3ec47604e4cc34e3dff1998f9a67cfbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 20:43:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"291241ca1af258b0a9a1c6ef480cad1bc11dddce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 20:53:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6f8714f52c3d41c38c444ebd9f44ca42 : SUCCESS in 7m 46s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/993faae2688847c392d64252d3dfb566 : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":1},{"id":"745820dec48c59c5a957b08ed9138ef4c4ad84d0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 20:53:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6378298a681dea4beea5b54aa0c05db9ce612bc2","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 21:05:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/213c96bba5b14049b4056be6ae1d8992 : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c12ac287ccbb38d662da3af56bae2d13083c60cf","revisions":{"c12ac287ccbb38d662da3af56bae2d13083c60cf":{"kind":"REWORK","_number":1,"created":"2021-03-16 21:56:58.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/84/780984/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/84/780984/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/84/780984/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/84/780984/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/84/780984/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/84/780984/1"}}},"commit":{"parents":[{"commit":"2bc6b1e5d9c7018418993a7273f579141f012be6","subject":"Flip nl02-04.openstack.org to nl02-04.opendev.org","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2bc6b1e5d9c7018418993a7273f579141f012be6"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2021-03-16 21:50:53.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2021-03-16 21:50:53.000000000","tz":-420},"subject":"Remove unused nl0X.openstack.org config files","message":"Remove unused nl0X.openstack.org config files\n\nOnce the opendev launchers are handling these duties and these servers\nhave all been removed from the system-config inventory we can go ahead\nand land this change to clean up the unused config files.\n\nChange-Id: I9792620eea81a07b6cbbfee37c08807114d2b390\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c12ac287ccbb38d662da3af56bae2d13083c60cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c12ac287ccbb38d662da3af56bae2d13083c60cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
