)]}'
{"id":"openstack%2Fproject-config~785769","triplet_id":"openstack%2Fproject-config~master~If26164513a34b4b3d1a68577c72b61f59444d232","project":"openstack/project-config","branch":"master","topic":"vexxhost-labels","hashtags":[],"change_id":"If26164513a34b4b3d1a68577c72b61f59444d232","subject":"Don\u0027t use standard labels in vexxhost-ca-ymq-1","status":"MERGED","created":"2021-04-11 15:10:10.000000000","updated":"2021-06-01 22:29:00.000000000","submitted":"2021-06-01 22:11:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"785769-vexxhost-labels","meta_rev_id":"ab40df20f97ffa106a8a984f650ffe5fe0004640","_number":785769,"virtual_id_number":785769,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-01 22:11:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2021-06-01 19:22:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2021-06-01 19:31:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-01 21:58:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-11 15:16:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-01 19:22:55.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2021-06-01 19:31:48.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"a65f2dc407f1cdd97c9e9f9ea58939dee8463c97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-04-11 15:10:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"357fa37b552571e1a4ab83cdf77756623fd88349","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-04-11 15:11:55.000000000","message":"Set Work In Progress\n\nRelated mailing list discussion: http://lists.opendev.org/pipermail/service-discuss/2021-April/000212.html","accounts_in_message":[],"_revision_number":1},{"id":"0748d96e1f3c3fe878eadbdb704161ecefd57819","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-04-11 15:16:50.000000000","message":"Patch Set 1:\n\nI left the nested-virt variety in here for now even though it\u0027s using the same flavor and we use the same label in other providers with only 8GB RAM, but it\u0027s worth discussing whether we want similar requirements for such \"nonstandard\" labels.\n\nIt also might make sense to also rebalance the quotas when changing this, rename the main pool, et cetera but I didn\u0027t want to muddy the initial proposal.","accounts_in_message":[],"_revision_number":1},{"id":"5c4dc7aad28b7ebf5aa2e189acf4d9aa3d12c0e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-11 15:16:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d06b9df95a704d109155c8b3a8f1a79d : SUCCESS in 5m 47s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/726f242c9b01456cb062f54d5ff956ce : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":1},{"id":"70453154a60ea2baee2dde2d91f68b27f74507ea","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-25 19:45:10.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"7d6690a264cb54714958b83d9f810ed73f5b0ad0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-25 19:51:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3b3e0d52a66f47448497a52f8be98c1947934d35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 19:58:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/cde5f50859ed4e138d304b2b38915223 : SUCCESS in 6m 16s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/b4fbf523996a48eb8d74dc8d4621ce4f : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":2},{"id":"26579bd59b73e9de1444cc1f80eb49b02a8e5d58","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-06-01 19:22:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"30988c97bfe3ed5af5cb29f6299cceadf45b999f","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-06-01 19:24:40.000000000","message":"Patch Set 2:\n\nFungi is looking for broad consensus before merging this change. I think the risk here is relatively low since we have other clouds providing the generic resources. I am happy to land this change and see what we learn. We can always revert it if necessary.","accounts_in_message":[],"_revision_number":2},{"id":"16f6eb651b8349de6df9264e3dc4a8822cf8c85e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-06-01 19:31:48.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"21ab184b13ab1f8ed6f16e11b0e4d83a3538a840","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-06-01 21:58:53.000000000","message":"Patch Set 2: Workflow+1\n\nApproving now as discussed in today\u0027s meeting.","accounts_in_message":[],"_revision_number":2},{"id":"f6bf92121936ec45833106269889b36a9e429301","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 21:59:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4c03f4fa0d69cebf925c245f1f886e20866ac5f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 22:11:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4575cdd5637342a8a0ae8841a0a67527 : SUCCESS in 6m 13s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/081aeb8c56654a4a95f874faca1b5ab9 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":2},{"id":"2b17802bc0bd348bd23038cea45a21253c721d0d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 22:11:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ab40df20f97ffa106a8a984f650ffe5fe0004640","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 22:29:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/13333a39a628438b87b712004ec77cd4 : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7f129bf7fc8e96b998310c3ffb93d95585316d8a","revisions":{"84cb82bce7262363bbc98c67bcd54a9461e3b428":{"kind":"REWORK","_number":1,"created":"2021-04-11 15:10:10.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/69/785769/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/69/785769/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/785769/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/785769/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/785769/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/69/785769/1"}}},"commit":{"parents":[{"commit":"93d20aee0e02ced6bcf5e0956f39cbb16994ed8d","subject":"Zero the min-ready for arm64 node labels","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/93d20aee0e02ced6bcf5e0956f39cbb16994ed8d"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-04-11 14:59:19.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-04-11 15:08:54.000000000","tz":0},"subject":"Don\u0027t use standard labels in vexxhost-ca-ymq-1","message":"Don\u0027t use standard labels in vexxhost-ca-ymq-1\n\nChange Ie424e7d6236c00678d92b34ed33dafaf966180fa replaced our custom\nosf-highcpu-8 flavor in Vexxhost\u0027s MTL (ca-ymq-1) region with\nv3-standard-8 which provides 32GB RAM instead of our usual 8GB. When\nmixed into our pools of standard node labels, this allows changes to\nmerge which cause some builds to consume far more RAM than will be\nable to run in our other providers, effectively breaking jobs for\nprojects where this happens.\n\nIn order to be able to continue to take advantage of the generous\nquota there, remove our standard node labels but still boot the\nspecial nested-virt, expanded, and gpu nodes which this provider\noffers.\n\nChange-Id: If26164513a34b4b3d1a68577c72b61f59444d232\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/84cb82bce7262363bbc98c67bcd54a9461e3b428"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/84cb82bce7262363bbc98c67bcd54a9461e3b428"}]},"branch":"refs/heads/master"},"7f129bf7fc8e96b998310c3ffb93d95585316d8a":{"kind":"REWORK","_number":2,"created":"2021-05-25 19:51:20.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/69/785769/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/69/785769/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/785769/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/785769/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/785769/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/69/785769/2"}}},"commit":{"parents":[{"commit":"1834b302966c737a36159407eeb21b46e44a12b3","subject":"Remove gerritbot from channels with no discussion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1834b302966c737a36159407eeb21b46e44a12b3"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-04-11 14:59:19.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-25 19:51:18.000000000","tz":0},"subject":"Don\u0027t use standard labels in vexxhost-ca-ymq-1","message":"Don\u0027t use standard labels in vexxhost-ca-ymq-1\n\nChange Ie424e7d6236c00678d92b34ed33dafaf966180fa replaced our custom\nosf-highcpu-8 flavor in Vexxhost\u0027s MTL (ca-ymq-1) region with\nv3-standard-8 which provides 32GB RAM instead of our usual 8GB. When\nmixed into our pools of standard node labels, this allows changes to\nmerge which cause some builds to consume far more RAM than will be\nable to run in our other providers, effectively breaking jobs for\nprojects where this happens.\n\nIn order to be able to continue to take advantage of the generous\nquota there, remove our standard node labels but still boot the\nspecial nested-virt, expanded, and gpu nodes which this provider\noffers.\n\nChange-Id: If26164513a34b4b3d1a68577c72b61f59444d232\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7f129bf7fc8e96b998310c3ffb93d95585316d8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7f129bf7fc8e96b998310c3ffb93d95585316d8a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
