)]}'
{"id":"openstack%2Fproject-config~786008","triplet_id":"openstack%2Fproject-config~master~I99f88e7b1f5049b39f21f74ea337cb549f3f8a95","project":"openstack/project-config","branch":"master","topic":"neutron-stable","hashtags":[],"change_id":"I99f88e7b1f5049b39f21f74ea337cb549f3f8a95","subject":"Update neutron stable grafana dashboards","status":"MERGED","created":"2021-04-13 07:23:24.000000000","updated":"2021-04-13 11:42:52.000000000","submitted":"2021-04-13 07:46:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"786008-neutron-stable","meta_rev_id":"c6391300cc11dd4c5933c82957e9b303f7b29d5d","_number":786008,"virtual_id_number":786008,"owner":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-13 07:46:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-13 07:37:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-13 07:37:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-13 07:30:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-13 07:37:21.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"}],"messages":[{"id":"b1da45c6ba7ed1867fb8175d26731fdec55ed28f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-04-13 07:23:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d52d050cd9d9eceacae7ae8cb0cd842ebd244bad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 07:30:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e75b35bd661445818cc7ace4a503de3c : SUCCESS in 5m 43s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/f44e791806f3426ebec3693bbde23ed4 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"c5659122499b094b742c70321eb9e7c3a777830a","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-04-13 07:37:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7a924fa4bc7eee1709e9c580c955048ee9d06470","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 07:37:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a630f3a1c94413f3417367d6a1b0237cef1f83be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 07:46:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d2b6b247410944419c75e7b1f4817326 : SUCCESS in 6m 37s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/e75a6ccd4a7f4049ae25cd139dfa51f3 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":1},{"id":"29eef304f1d297fd021116652250253788373715","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 07:46:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c6391300cc11dd4c5933c82957e9b303f7b29d5d","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 11:42:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-grafana https://zuul.opendev.org/t/openstack/build/edfbffbbfc3845f695b87735681dbb5f : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2d05ad49ddeab8fccc31828caba9327429061a99","revisions":{"2d05ad49ddeab8fccc31828caba9327429061a99":{"kind":"REWORK","_number":1,"created":"2021-04-13 07:23:24.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/08/786008/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/08/786008/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/786008/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/786008/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/786008/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/08/786008/1"}}},"commit":{"parents":[{"commit":"9ce1291e71e3c8957639374fb56e9b8c43e9731e","subject":"Merge \"nodepool elements: create suse boot rc directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9ce1291e71e3c8957639374fb56e9b8c43e9731e"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2021-04-13 07:13:41.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2021-04-13 07:22:40.000000000","tz":120},"subject":"Update neutron stable grafana dashboards","message":"Update neutron stable grafana dashboards\n\nWith wallaby release, it is time again to rotate the stable dashboards.\nWallaby one was synced from master dashboard, adjusting to stable\nbranch, neutron-tempest-plugin job names and removing -master jobs.\n\nChange-Id: I99f88e7b1f5049b39f21f74ea337cb549f3f8a95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2d05ad49ddeab8fccc31828caba9327429061a99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2d05ad49ddeab8fccc31828caba9327429061a99"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
