)]}'
{"id":"openstack%2Fproject-config~786773","triplet_id":"openstack%2Fproject-config~master~Id971e42edc1102638b90e58cc4ae43d0120b31dc","project":"openstack/project-config","branch":"master","topic":"debian-bullseye","hashtags":[],"change_id":"Id971e42edc1102638b90e58cc4ae43d0120b31dc","subject":"Add Debian Bullseye.","status":"ABANDONED","created":"2021-04-18 16:44:57.000000000","updated":"2021-04-20 19:39:08.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5d845baf1c10713da0a4aaaeec98f506ca34b15e","_number":786773,"virtual_id_number":786773,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-18 16:54:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-18 16:54:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-18 17:31:12.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2021-04-18 18:36:49.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"}],"messages":[{"id":"6e319e5bab06e18f71a6c0a51cb7bacd0bf7967a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-18 16:44:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"029431d0500482238397cc1f4949321667100f9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-18 16:54:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2fef9b284fd54eb09aefeca0b2c037ff : SUCCESS in 8m 41s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/9bd87bbdedd147fcb15e865fc9a16ee7 : SUCCESS in 3m 50s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/6eb4cc494b31469aaa26c2ed3f40f284 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":1},{"id":"88b16284d4d8fcbbe05b049545bc307f8819cf42","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-18 17:31:12.000000000","message":"Patch Set 1:\n\nHow does it differ from https://review.opendev.org/c/openstack/project-config/+/783613 one?","accounts_in_message":[],"_revision_number":1},{"id":"60dc25170c7ba3be7dcff604436d9a94234b0468","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-04-18 18:36:49.000000000","message":"Patch Set 1:\n\nI think this is a duplicate of I9c1bb7aaa02ba60ee52e2d7b990e2e6e1212317f.","accounts_in_message":[],"_revision_number":1},{"id":"e7dc41f0d9bbe998f59428108fd4e4e6dec82507","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-18 23:02:46.000000000","message":"Patch Set 1:\n\nOk, then I abandon this one.","accounts_in_message":[],"_revision_number":1},{"id":"5d845baf1c10713da0a4aaaeec98f506ca34b15e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-20 19:39:08.000000000","message":"Abandoned\n\nsuperceded","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c253e8d7cbacfa728cd58244fd203739eb78c1b1","revisions":{"c253e8d7cbacfa728cd58244fd203739eb78c1b1":{"kind":"REWORK","_number":1,"created":"2021-04-18 16:44:57.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/73/786773/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/73/786773/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/786773/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/786773/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/786773/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/73/786773/1"}}},"commit":{"parents":[{"commit":"f4fac9a53eee5c4bd9462bd007c578073a6bb483","subject":"ACL, allow merge for airship-release group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f4fac9a53eee5c4bd9462bd007c578073a6bb483"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-18 16:44:36.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-18 16:44:36.000000000","tz":120},"subject":"Add Debian Bullseye.","message":"Add Debian Bullseye.\n\nChange-Id: Id971e42edc1102638b90e58cc4ae43d0120b31dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c253e8d7cbacfa728cd58244fd203739eb78c1b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c253e8d7cbacfa728cd58244fd203739eb78c1b1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
