)]}'
{"id":"openstack%2Fproject-config~787523","triplet_id":"openstack%2Fproject-config~master~I2fd7a6387d2f50eeeb8cef513df19b5696cce55b","project":"openstack/project-config","branch":"master","topic":"review-priority","hashtags":[],"change_id":"I2fd7a6387d2f50eeeb8cef513df19b5696cce55b","subject":"Add review priority label to nova deliverables","status":"MERGED","created":"2021-04-22 10:19:12.000000000","updated":"2021-08-27 16:41:39.000000000","submitted":"2021-08-27 16:23:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":9,"has_review_started":true,"submission_id":"787523-review-priority","meta_rev_id":"a42e9ac3190ca5bc5d716bcfeda33a9a3dcc291f","_number":787523,"virtual_id_number":787523,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-27 16:23:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":2,"date":"2021-08-27 16:03:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":1,"date":"2021-08-27 16:03:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-22 10:20:17.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2021-04-22 10:26:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-29 16:24:46.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2021-04-29 16:59:07.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2021-05-07 09:16:05.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2021-05-20 15:44:18.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-06-01 11:39:54.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-08-27 12:29:00.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2021-08-27 15:15:02.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"bd9d7fee910ea424db7c227168458510e2b22e40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-04-22 10:19:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0c82ceeb1ce820e336e8301b63f93b33843b6c6","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-04-22 10:20:17.000000000","message":"Patch Set 1:\n\nadding gibi as nova ptl","accounts_in_message":[],"_revision_number":1},{"id":"c375da88b700c6a5ce028c0f5dc350610c431a4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 10:26:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/78c44a50a5444a3cb3acd9a07af565df : SUCCESS in 6m 21s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/dadd6f526bd3436688dfd54e33a5c018 : FAILURE in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"a8433d63f865d2fe4f20284e92a2e91772aa4ae6","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-04-22 11:49:11.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nI\u0027m wondering about a couple of things:\n\n* The current proposal (based on other projects\u0027 usage) introduces 4 values (-1, 0, +1, +2) for this new label. Right now I don\u0027t think we need such amount of complexity. I\u0027m wondering if we are OK with a simple 0,+1 flag. Simply +1 would mean that the core team (or some of the cores) is decided to put extra attention to the given patch. \n\n* The current proposal allows every core to set the flag. I like this. I expect that the core who sets this flag commits to review the marked patch. (This also prevent setting the mark on its own patch)\n\n* Similar to the runway process here I also would like to limit the amount of stuff that can have priority at the same time. So I suggest that before you set the flag check if there are other patches already marked where you can help with the review and if so, then do that review first, before you mark a new patch as priority.\n\n* I will add an agenda item to the weekly meeting where we look at the marked patches. So that we can have a chat about the status or decided to remove the mark if the patch stalled out.\n\nWhat do you think?","accounts_in_message":[],"_revision_number":1},{"id":"02205a32d7b58df4b84e0ea9a8bb9ea1774dcbc7","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-04-22 13:37:06.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e I\u0027m wondering about a couple of things:\n\u003e \n\u003e * The current proposal (based on other projects\u0027 usage) introduces 4 values (-1, 0, +1, +2) for this new label. Right now I don\u0027t think we need such amount of complexity. I\u0027m wondering if we are OK with a simple 0,+1 flag. Simply +1 would mean that the core team (or some of the cores) is decided to put extra attention to the given patch.\n\nya so i was orginally going to make it 0,+1 and +2 because if we dont like semthign we can alwasy -2 the idea so negitive values seamed weird to me.\n\n0 an +1 i think would be fine too\n \n\u003e \n\u003e * The current proposal allows every core to set the flag. I like this. I expect that the core who sets this flag commits to review the marked patch. (This also prevent setting the mark on its own patch)\n\nyes so just like code reviews when you leave a +p( i cant keep spelling priority right :P) \nit will put your name beside it in the ui so we can clearly see that in the ui and know who added it\n\nmultiple cores can set it like +w\n\n\u003e \n\u003e * Similar to the runway process here I also would like to limit the amount of stuff that can have priority at the same time. So I suggest that before you set the flag check if there are other patches already marked where you can help with the review and if so, then do that review first, before you mark a new patch as priority.\n\nthat sounds resonable altough that is something not enforcable by gerrit :) \n\nsetting -w until agreement is reached on the questions above.","accounts_in_message":[],"_revision_number":1},{"id":"1a98147b3202adf48baf557d7751cccd9fc5c2b9","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-04-28 23:42:41.000000000","message":"Patch Set 1:\n\nlook like no one else has express an opinion.\nill bring this up in the nova team meeting this week.\nif there is no other input ill just update this to what gibi has suggested as i generally\nagree with his suggestions.","accounts_in_message":[],"_revision_number":1},{"id":"ef0206632fbe0fbf6fd9bdc77b88df1a2f5866ae","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-04-29 16:24:46.000000000","message":"Patch Set 1:\n\nI agree with 0,1 suggestion which will make signal a clear instead of interpreting -2,-1.0,+1,+2","accounts_in_message":[],"_revision_number":1},{"id":"0cca22a434fcd6f4e0c17651599f1825154eb22a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-04-29 16:59:07.000000000","message":"Patch Set 1:\n\n(4 comments)\n\n* I agree: 0..+1 is enough\n* I could imagine that nova-core members could add priority to stable patches but as it was said on the meeting, the nova-core and nova-stable-maint overlap maybe covers this","accounts_in_message":[],"_revision_number":1},{"id":"2c3119ac4ca7293b56540687a500c77bdbcb4fcf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-05-06 13:06:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9136af63de186fced8c74788627baddf0624b258","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 13:13:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4cfae421b6c84a588582f2220d8bf9e9 : SUCCESS in 5m 31s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/ef01f321a9174eeab6350d1ece346c92 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":2},{"id":"88fd20ecb272b1ba93d31ef97cd382766fa40822","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-05-06 17:01:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d1cfee6b7b8c2e25e83c6439ae66b231ca5a2a4d","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2021-05-07 09:16:05.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5dd4f246c813680bf500177389647751e55fa9a4","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2021-05-07 09:28:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"02aa2442b9acd2380abaa26eaa8635a29b651f29","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-05-07 12:41:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"00fa0fbf2eb12f6ba082030869c697e01c931bda","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-05-10 15:33:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6c7a1d631183c770980855a38bf6502ec2de3943","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2021-05-11 15:54:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7f4604bd0590c92787e782d0f2bdb98f8792f18b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-05-11 17:17:27.000000000","message":"Patch Set 2: Workflow-1\n\nthis need to be defined inlien\n\nhttps://opendev.org/openstack/project-config/src/branch/master/gerrit/acls/openstack/cinder.config#L25-L35\n\nand we have not decided on the process so -w untill that is decided","accounts_in_message":[],"_revision_number":2},{"id":"150f90ac3d4ce98b1c0e36094b29b7ce2595fa14","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-05-20 11:29:07.000000000","message":"Patch Set 2:\n\n(1 comment)\n\n\u003e Patch Set 2: Workflow-1\n\u003e \n\u003e this need to be defined inlien\n\u003e \n\u003e https://opendev.org/openstack/project-config/src/branch/master/gerrit/acls/openstack/cinder.config#L25-L35\n\u003e \n\u003e and we have not decided on the process so -w untill that is decided","accounts_in_message":[],"_revision_number":2},{"id":"c7e7a95a61429f39c6b5cd925c4abac6800c8c25","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-05-20 15:44:18.000000000","message":"Patch Set 2: Code-Review+1\n\nI am fine with the current proposal and if this does not work or need to be extended then we can always do.","accounts_in_message":[],"_revision_number":2},{"id":"c41f531ba4d211b53f0f08d007ba9d0ba97e650d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-06-01 11:39:54.000000000","message":"Patch Set 2: Code-Review+1\n\nBased on gibi\u0027s contributor doc changes, I think this makes sense as-is","accounts_in_message":[],"_revision_number":2},{"id":"428121b58e01f387e972e081b131eec993a9ecc2","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-06-01 14:11:44.000000000","message":"Patch Set 2:\n\nmy implemantion is still correct since i need to define the label in each of the configs\nsince its not global. but if there is a concenos on the contibutors guide now ill rework this to reflect that. i had been waiting of that before revising this but ill do that shortly.","accounts_in_message":[],"_revision_number":2},{"id":"cbacdb5b19ce7d63bca474db989a30dbf34d31a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-08-23 12:46:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"875fec650ef2e6c9d1b213ccd3b20a727701499c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-08-23 12:47:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"eb07d42f67e134a38a26d438437328d041151cdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 13:00:38.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b6be6cd7b4d342eab81037010209edf8 : SUCCESS in 6m 52s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/27387f3d986644ccabe1d0631e95d35a : FAILURE in 4m 22s","accounts_in_message":[],"_revision_number":4},{"id":"4f0a02334b7c01231a0259da9ed2a55937790c3c","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-08-23 13:13:45.000000000","message":"Patch Set 4:\n\nlooks like normalize acls dose not know that https://review.opendev.org/Documentation/config-labels.html#label_copyAnyScore is a thing","accounts_in_message":[],"_revision_number":4},{"id":"83b894cf3723a96cee3dc4c6a749f15acf4ef9ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-08-23 13:20:12.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"aa652150564bb7298398bef8d5a5492f3005630e","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-08-23 13:29:41.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"454c616d9b034288f74a0b68e0472bc6a43d0d0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-08-23 13:34:31.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"20ebd3ec710b8e1bb4c1df3f7a5d09b2349e0f8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 13:45:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8c1e5d5313f74ff2865882ff149642a4 : SUCCESS in 7m 13s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8bc891a1634c4bc4a38affed5e354188 : SUCCESS in 7m 43s","accounts_in_message":[],"_revision_number":6},{"id":"dc792eeadaacd7f6a48bf39594ad85cef742c044","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-08-23 17:14:19.000000000","message":"Patch Set 6: Code-Review+1\n\nThanks Sean, looks good to me.","accounts_in_message":[],"_revision_number":6},{"id":"3d91017321e96c08bef0a2c866c3c2fa93116618","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-08-27 12:29:00.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2797386563e7033df4fcf71ffd1a236c8b20a8c5","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-08-27 15:15:02.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)\n\nFeel free to treat this as a soft -1 but I think the suggested inline cleanup would be good to do on all of the acl files.","accounts_in_message":[],"_revision_number":6},{"id":"798463531cb65bf9d7332c0b325d4b50f263e526","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-08-27 16:02:24.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"57f3b48d798f423bcb3989c47de3292cad858fcd","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-08-27 16:03:44.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"8e74d1e57151faf29b2e23e392f1138d8c0e5c4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 16:12:16.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/673343deb0b64bdca2f3bbf24715c431 : SUCCESS in 7m 53s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/9ec1a182210d4e608d45b57af6eb7e29 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":7},{"id":"f1b5a772586a427192b4ec2123b261ba2d9632b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 16:12:39.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"ddeb2d04909b38666d078ba49cf0ecb8c37be526","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 16:23:13.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/940e8e56f4df43d6b7cd6edfd1d9d693 : SUCCESS in 6m 34s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/407a1e7556f54623a2faf2dc1849e225 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":7},{"id":"0540c08fd745db23f4a7d43f637cec5dffb7c687","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 16:23:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"a42e9ac3190ca5bc5d716bcfeda33a9a3dcc291f","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 16:41:39.000000000","message":"Patch Set 7:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/c2652967c481476f86f04436814643ab : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"d41dfcd16f77de196af7d8a17673b4718fac9257","revisions":{"4dde4b3a211a86e3e25b4af0058b80aece3c2182":{"kind":"REWORK","_number":1,"created":"2021-04-22 10:19:12.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/23/787523/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/23/787523/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/787523/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/787523/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/787523/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/23/787523/1"}}},"commit":{"parents":[{"commit":"36542633cf98782e4a5e719342955a425c61dfd7","subject":"Change cyborg project track to launchpad","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/36542633cf98782e4a5e719342955a425c61dfd7"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2021-04-22 10:07:24.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2021-04-22 10:18:03.000000000","tz":60},"subject":"Add review priority label to nova deliverables","message":"Add review priority label to nova deliverables\n\nThis change adds a Review Priority label to all nova deliverables\ncurrently under acl control in the project config repo.\n\nThe range -1..+2 was chosen to have parity with all\nexisting uses fo the label for consitency.\n\nThe ablity set the new lable is granted only to the core\nand stable core teams for the updated repos.\n\nChange-Id: I2fd7a6387d2f50eeeb8cef513df19b5696cce55b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4dde4b3a211a86e3e25b4af0058b80aece3c2182"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4dde4b3a211a86e3e25b4af0058b80aece3c2182"}]},"branch":"refs/heads/master"},"102aeb475085e3d97d76f72bae5dfa988b90547d":{"kind":"REWORK","_number":2,"created":"2021-05-06 13:06:28.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/23/787523/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/23/787523/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/787523/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/787523/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/787523/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/23/787523/2"}}},"commit":{"parents":[{"commit":"36542633cf98782e4a5e719342955a425c61dfd7","subject":"Change cyborg project track to launchpad","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/36542633cf98782e4a5e719342955a425c61dfd7"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2021-04-22 10:07:24.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2021-05-06 13:04:48.000000000","tz":60},"subject":"Add review priority label to nova deliverables","message":"Add review priority label to nova deliverables\n\nThis change adds a Review Priority label to all nova deliverables\ncurrently under acl control in the project config repo.\n\nThe ability set the new label is granted only to the core\nand stable core teams for the updated repos.\n\nChange-Id: I2fd7a6387d2f50eeeb8cef513df19b5696cce55b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/102aeb475085e3d97d76f72bae5dfa988b90547d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/102aeb475085e3d97d76f72bae5dfa988b90547d"}]},"branch":"refs/heads/master"},"58dd927f0ef77607326874f3d4fe523765905fbe":{"kind":"REWORK","_number":3,"created":"2021-08-23 12:46:22.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/23/787523/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/23/787523/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/787523/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/787523/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/787523/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/23/787523/3"}}},"commit":{"parents":[{"commit":"36542633cf98782e4a5e719342955a425c61dfd7","subject":"Change cyborg project track to launchpad","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/36542633cf98782e4a5e719342955a425c61dfd7"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2021-04-22 10:07:24.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2021-08-23 12:46:06.000000000","tz":60},"subject":"Add review priority label to nova deliverables","message":"Add review priority label to nova deliverables\n\nThis change adds a Review Priority label to all nova deliverables\ncurrently under acl control in the project config repo.\n\nThe ability set the new label is granted only to the core\nand stable core teams for the updated repos.\n\nChange-Id: I2fd7a6387d2f50eeeb8cef513df19b5696cce55b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/58dd927f0ef77607326874f3d4fe523765905fbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/58dd927f0ef77607326874f3d4fe523765905fbe"}]},"branch":"refs/heads/master"},"7bd2e16ccedb5986f49cd1654f259fb77727ed1a":{"kind":"REWORK","_number":4,"created":"2021-08-23 12:47:12.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/23/787523/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/23/787523/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/787523/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/787523/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/787523/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/23/787523/4"}}},"commit":{"parents":[{"commit":"36542633cf98782e4a5e719342955a425c61dfd7","subject":"Change cyborg project track to launchpad","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/36542633cf98782e4a5e719342955a425c61dfd7"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2021-04-22 10:07:24.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2021-08-23 12:46:41.000000000","tz":60},"subject":"Add review priority label to nova deliverables","message":"Add review priority label to nova deliverables\n\nThis change adds a Review Priority label to all nova deliverables\ncurrently under acl control in the project config repo.\n\nThe ability set the new label is granted only to the core\nand stable core teams for the updated repos.\n\nChange-Id: I2fd7a6387d2f50eeeb8cef513df19b5696cce55b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7bd2e16ccedb5986f49cd1654f259fb77727ed1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7bd2e16ccedb5986f49cd1654f259fb77727ed1a"}]},"branch":"refs/heads/master"},"6e6b43b860c8b73a7331bda1c4e4651fa510f7fa":{"kind":"REWORK","_number":5,"created":"2021-08-23 13:20:12.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/23/787523/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/23/787523/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/787523/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/787523/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/787523/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/23/787523/5"}}},"commit":{"parents":[{"commit":"36542633cf98782e4a5e719342955a425c61dfd7","subject":"Change cyborg project track to launchpad","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/36542633cf98782e4a5e719342955a425c61dfd7"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2021-04-22 10:07:24.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2021-08-23 13:20:04.000000000","tz":60},"subject":"Add review priority label to nova deliverables","message":"Add review priority label to nova deliverables\n\nThis change adds a Review Priority label to all nova deliverables\ncurrently under acl control in the project config repo.\n\nThe ability set the new label is granted only to the core\nand stable core teams for the updated repos.\n\nChange-Id: I2fd7a6387d2f50eeeb8cef513df19b5696cce55b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6e6b43b860c8b73a7331bda1c4e4651fa510f7fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6e6b43b860c8b73a7331bda1c4e4651fa510f7fa"}]},"branch":"refs/heads/master"},"1e8e9d9719b2b4e89401a295ff5473d78ece68f3":{"kind":"REWORK","_number":6,"created":"2021-08-23 13:34:31.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/23/787523/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/23/787523/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/787523/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/787523/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/787523/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/23/787523/6"}}},"commit":{"parents":[{"commit":"36542633cf98782e4a5e719342955a425c61dfd7","subject":"Change cyborg project track to launchpad","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/36542633cf98782e4a5e719342955a425c61dfd7"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2021-04-22 10:07:24.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2021-08-23 13:34:24.000000000","tz":60},"subject":"Add review priority label to nova deliverables","message":"Add review priority label to nova deliverables\n\nThis change adds a Review Priority label to all nova deliverables\ncurrently under acl control in the project config repo.\n\nThe ability set the new label is granted only to the core\nand stable core teams for the updated repos.\n\nChange-Id: I2fd7a6387d2f50eeeb8cef513df19b5696cce55b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1e8e9d9719b2b4e89401a295ff5473d78ece68f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1e8e9d9719b2b4e89401a295ff5473d78ece68f3"}]},"branch":"refs/heads/master"},"d41dfcd16f77de196af7d8a17673b4718fac9257":{"kind":"REWORK","_number":7,"created":"2021-08-27 16:02:24.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/23/787523/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/23/787523/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/787523/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/787523/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/787523/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/23/787523/7"}}},"commit":{"parents":[{"commit":"36542633cf98782e4a5e719342955a425c61dfd7","subject":"Change cyborg project track to launchpad","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/36542633cf98782e4a5e719342955a425c61dfd7"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2021-04-22 10:07:24.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2021-08-27 16:02:16.000000000","tz":60},"subject":"Add review priority label to nova deliverables","message":"Add review priority label to nova deliverables\n\nThis change adds a Review Priority label to all nova deliverables\ncurrently under acl control in the project config repo.\n\nThe ability set the new label is granted only to the core\nand stable core teams for the updated repos.\n\nChange-Id: I2fd7a6387d2f50eeeb8cef513df19b5696cce55b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d41dfcd16f77de196af7d8a17673b4718fac9257"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d41dfcd16f77de196af7d8a17673b4718fac9257"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
