)]}'
{"id":"openstack%2Fproject-config~787524","triplet_id":"openstack%2Fproject-config~master~Ia43306e47dc9e70f3b26593765210abfbd8ec6fe","project":"openstack/project-config","branch":"master","topic":"review-priority","hashtags":[],"change_id":"Ia43306e47dc9e70f3b26593765210abfbd8ec6fe","subject":"Copy placement acls to all placement repos","status":"ABANDONED","created":"2021-04-22 10:19:12.000000000","updated":"2021-04-22 13:30:42.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8e876d883d6e61f990c57d563e7eb86d7342b8c6","_number":787524,"virtual_id_number":787524,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-04-22 10:32:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-22 10:20:50.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2021-04-22 10:32:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-22 11:59:49.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"}],"messages":[{"id":"db562c9744e28c9a4639605ad8141579d5d84962","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-04-22 10:19:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f23a1234454da7541fc69826d5f26e249892448b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-04-22 10:20:50.000000000","message":"Patch Set 1:\n\nadding gibi as placement ptl","accounts_in_message":[],"_revision_number":1},{"id":"e9f555d851fabc3a79fa2b256774689efa061212","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 10:32:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/bc5f8816aef34e90bed7d72e835513da : SUCCESS in 6m 12s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/cf40e962be9b4edeb5098ddd8fca9155 : FAILURE in 4m 51s","accounts_in_message":[],"_revision_number":1},{"id":"0f0c74115cf2cd9c8f12cedd8f7e77b8097a1f8f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-04-22 11:59:49.000000000","message":"Patch Set 1:\n\nSee gerrit/projects.yaml, these repositories all share the placement ACL file.","accounts_in_message":[],"_revision_number":1},{"id":"8e876d883d6e61f990c57d563e7eb86d7342b8c6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-04-22 13:30:42.000000000","message":"Abandoned\n\nah in that case then i dont need to copy it the current logic will have the same effect. thanks for pointing that out.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4f1a5365f7b907375a77a8f22532048733f6c7ea","revisions":{"4f1a5365f7b907375a77a8f22532048733f6c7ea":{"kind":"REWORK","_number":1,"created":"2021-04-22 10:19:12.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/24/787524/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/24/787524/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/787524/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/787524/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/787524/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/24/787524/1"}}},"commit":{"parents":[{"commit":"4dde4b3a211a86e3e25b4af0058b80aece3c2182","subject":"Add review priority label to nova deliverables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4dde4b3a211a86e3e25b4af0058b80aece3c2182"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2021-04-22 10:13:18.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2021-04-22 10:18:56.000000000","tz":60},"subject":"Copy placement acls to all placement repos","message":"Copy placement acls to all placement repos\n\nThis adds the same acl configuration used by the main placement\nrepo to its related projects. namely os-resource-classes,\nos-traits and osc-placement.\n\nChange-Id: Ia43306e47dc9e70f3b26593765210abfbd8ec6fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4f1a5365f7b907375a77a8f22532048733f6c7ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4f1a5365f7b907375a77a8f22532048733f6c7ea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
