)]}'
{"id":"openstack%2Fproject-config~788556","triplet_id":"openstack%2Fproject-config~master~I4a66c22fa66056b501e115f013ac545223e0e96a","project":"openstack/project-config","branch":"master","topic":"openstack-releases","hashtags":[],"change_id":"I4a66c22fa66056b501e115f013ac545223e0e96a","subject":"Move projects under meta-config acl (5)","status":"MERGED","created":"2021-04-28 17:28:00.000000000","updated":"2021-04-29 22:14:03.000000000","submitted":"2021-04-29 21:34:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"788556-openstack-releases","meta_rev_id":"84c305c6f014f8643230dd1e32a792700a3ab79c","_number":788556,"virtual_id_number":788556,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-29 21:34:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2021-04-29 22:14:03.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2021-04-29 08:40:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-29 21:20:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-29 21:20:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-28 17:39:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-28 17:53:15.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2021-04-29 08:40:49.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-04-29 21:20:25.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"33203f06269cec858536e25396f3cae41a7ba886","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-04-28 17:28:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"17493406950ef3c120a24dcc5215ba64e75b9c34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 17:39:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f5609c33d1ce4a7b9c86183533216a7c : SUCCESS in 5m 35s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/637f6c4fad884b44817b2f47ae10cec5 : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":1},{"id":"1ee379617b296849dd23e06ca041faa8f5269ed0","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-04-28 17:53:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"de503289ece0df05c56ac01bb942497700a75dcc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-04-28 19:40:31.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"9334f5170d8227f14e844c9336813ea9be3f9265","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 19:48:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/24c14bac4f1844ff8f4b3d764fd6876a : SUCCESS in 6m 03s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/939cccb244904d46840c71fd583a2584 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":2},{"id":"255eb711417baf423e6a4e990891286a11a774f1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-04-29 08:40:49.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"dbeab9f7de5e9a910569b86d16040691a9fc9502","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-04-29 21:20:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nApproving these in batches just so we don\u0027t cause a logjam.","accounts_in_message":[],"_revision_number":2},{"id":"43b357a2b233263e5b9ff1b27b0200422be27bbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 21:21:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fad3f895e391cf1d8913880916305b93e4079da8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 21:34:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7d05d55821c54b03a7e6e558f94142cc : SUCCESS in 5m 04s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/356e481040fa41af8b0aebb5077a5bee : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":2},{"id":"8eb57f04e338de3f474be467be369f99de9fb3d2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 21:34:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e5978bd0c56991cadf115f3626e6edc3a31fbe8b","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 22:14:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/0381cc9afa1449ab8779f2a1c39672d4 : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8f1c41d466513a1ba3fea8818d81b425af45a434","revisions":{"3bfaecf81aaf97c4caffc5548a356da451722cb9":{"kind":"REWORK","_number":1,"created":"2021-04-28 17:28:00.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/56/788556/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/56/788556/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/788556/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/788556/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/788556/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/56/788556/1"}}},"commit":{"parents":[{"commit":"72c9258de6dbf74786a4a687d61fdfee4cced448","subject":"Move projects under meta-config acl (4)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/72c9258de6dbf74786a4a687d61fdfee4cced448"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-04-28 17:26:30.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-04-28 17:26:30.000000000","tz":120},"subject":"Move projects under meta-config acl (5)","message":"Move projects under meta-config acl (5)\n\nThis patch series move openstack/* projects from All-Projects ACL to\nmeta-config ACL, which is created especially for the Release Managers\ngroup.\n\nChange-Id: I4a66c22fa66056b501e115f013ac545223e0e96a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3bfaecf81aaf97c4caffc5548a356da451722cb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3bfaecf81aaf97c4caffc5548a356da451722cb9"}]},"branch":"refs/heads/master"},"8f1c41d466513a1ba3fea8818d81b425af45a434":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-04-28 19:40:31.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/56/788556/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/56/788556/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/788556/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/788556/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/788556/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/56/788556/2"}}},"commit":{"parents":[{"commit":"6bd9d39d4d27d6c745c2c40231ddb8b3666db921","subject":"Adds r/stx.5.0 to the list of branches to promote for starlingx docs.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6bd9d39d4d27d6c745c2c40231ddb8b3666db921"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-04-28 17:26:30.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-04-28 19:40:31.000000000","tz":0},"subject":"Move projects under meta-config acl (5)","message":"Move projects under meta-config acl (5)\n\nThis patch series move openstack/* projects from All-Projects ACL to\nmeta-config ACL, which is created especially for the Release Managers\ngroup.\n\nChange-Id: I4a66c22fa66056b501e115f013ac545223e0e96a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8f1c41d466513a1ba3fea8818d81b425af45a434"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8f1c41d466513a1ba3fea8818d81b425af45a434"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
