)]}'
{"id":"openstack%2Fproject-config~788561","triplet_id":"openstack%2Fproject-config~master~Icaee2a8fce4d50891948c6d8351aea974d974963","project":"openstack/project-config","branch":"master","topic":"openstack-releases","hashtags":[],"change_id":"Icaee2a8fce4d50891948c6d8351aea974d974963","subject":"Move projects under meta-config acl (8)","status":"MERGED","created":"2021-04-28 17:49:00.000000000","updated":"2021-04-29 23:18:16.000000000","submitted":"2021-04-29 22:57:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"788561-openstack-releases","meta_rev_id":"e8d138784ff50b33311f86e9e7568a5c64090004","_number":788561,"virtual_id_number":788561,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-29 22:57:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2021-04-29 23:18:16.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2021-04-29 08:40:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-29 22:44:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-29 22:44:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-28 18:01:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-28 18:14:04.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2021-04-29 08:40:01.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-04-29 22:44:16.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"7ccb08506d005ca345608b55d7a5bf3c38a80190","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-04-28 17:49:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"41b661afa3263282fb6a42c2ef4350ae3d86d08c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 18:01:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/727fb5b3984f4f9eb95e0f70b70c0102 : SUCCESS in 7m 17s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/325277f692b147578b36a7cc8a87efff : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":1},{"id":"b9111a0d1a673d0fdcbcd952143aa3e444c8d022","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-04-28 18:14:04.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e2994e5715c47da6eb2ac9bcfcbf2883b7428267","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-04-28 19:44:16.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"1effb09565f27e3546977e3703f6fb3046c4d17d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 19:51:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e50224af2a0741da9cee3fabc3a6122b : SUCCESS in 5m 23s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/e2637bcb89034f8b9e6bed29c9b4a845 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":2},{"id":"400aa66592514fe310bc7526bb90c47a7847d2ee","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-04-29 08:40:01.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e9797a333845967e4cb4092c87765157b01a1296","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-04-29 22:44:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nApproving another subset.","accounts_in_message":[],"_revision_number":2},{"id":"7816108f1af89d628d866c35c2005ef6b8a1d1c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 22:44:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"47b3f709693db37be0fe8c2cbcb9fa5026966e54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 22:57:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9c53c531f41242a7bf936249ddc4f7a0 : SUCCESS in 6m 03s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8f23fc5564d94a909b0673deeeaa1051 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":2},{"id":"0b0f76e0d2affc50e1a99258ac5018e51a793bd5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 22:57:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"0434d67f59dea69a49a3c5fc785875df6f681ebc","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 23:18:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/35bd367c123c49c49a1d83ca01bae497 : SUCCESS in 8m 33s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ef0de80b7fe9acfe894438034bf7bf26ba993496","revisions":{"3b1c3181deea8375a1e92b1531eaa1221802c735":{"kind":"REWORK","_number":1,"created":"2021-04-28 17:49:00.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/61/788561/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/61/788561/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/61/788561/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/61/788561/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/61/788561/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/61/788561/1"}}},"commit":{"parents":[{"commit":"7bce3e84acd8a910da04f7ce0c31d6565bd4c934","subject":"Move projects under meta-config acl (7)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7bce3e84acd8a910da04f7ce0c31d6565bd4c934"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-04-28 17:47:56.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-04-28 17:47:56.000000000","tz":120},"subject":"Move projects under meta-config acl (8)","message":"Move projects under meta-config acl (8)\n\nThis patch series move openstack/* projects from All-Projects ACL to\nmeta-config ACL, which is created especially for the Release Managers\ngroup.\n\nChange-Id: Icaee2a8fce4d50891948c6d8351aea974d974963\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3b1c3181deea8375a1e92b1531eaa1221802c735"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3b1c3181deea8375a1e92b1531eaa1221802c735"}]},"branch":"refs/heads/master"},"ef0de80b7fe9acfe894438034bf7bf26ba993496":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-04-28 19:44:16.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/61/788561/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/61/788561/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/61/788561/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/61/788561/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/61/788561/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/61/788561/2"}}},"commit":{"parents":[{"commit":"6bd9d39d4d27d6c745c2c40231ddb8b3666db921","subject":"Adds r/stx.5.0 to the list of branches to promote for starlingx docs.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6bd9d39d4d27d6c745c2c40231ddb8b3666db921"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-04-28 17:47:56.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-04-28 19:44:16.000000000","tz":0},"subject":"Move projects under meta-config acl (8)","message":"Move projects under meta-config acl (8)\n\nThis patch series move openstack/* projects from All-Projects ACL to\nmeta-config ACL, which is created especially for the Release Managers\ngroup.\n\nChange-Id: Icaee2a8fce4d50891948c6d8351aea974d974963\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ef0de80b7fe9acfe894438034bf7bf26ba993496"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ef0de80b7fe9acfe894438034bf7bf26ba993496"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
