)]}'
{"id":"openstack%2Fproject-config~788562","triplet_id":"openstack%2Fproject-config~master~Ic1722c158d0f904cc7eb10bd09670725f3037dad","project":"openstack/project-config","branch":"master","topic":"starlingx-add-promote-branch","hashtags":[],"change_id":"Ic1722c158d0f904cc7eb10bd09670725f3037dad","subject":"Adds r/stx.5.0 to the list of branches to promote for starlingx docs.","status":"MERGED","created":"2021-04-28 17:52:22.000000000","updated":"2021-04-28 19:21:52.000000000","submitted":"2021-04-28 19:21:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"788562-starlingx-add-promote-branch","meta_rev_id":"175d810d3cce53e4616ed02c8d7a0deeb6dc93df","_number":788562,"virtual_id_number":788562,"owner":{"_account_id":33339,"name":"Kevin Putnam","email":"kevin.putnam@intel.com","username":"ikevinputnam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30859,"name":"M Camp","email":"mcamp859@hotmail.com","username":"mcamp859"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-28 19:21:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":30859,"name":"M Camp","email":"mcamp859@hotmail.com","username":"mcamp859"},"all":[{"value":1,"date":"2021-04-28 19:00:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30859,"name":"M Camp","email":"mcamp859@hotmail.com","username":"mcamp859"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-28 18:59:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":30859,"name":"M Camp","email":"mcamp859@hotmail.com","username":"mcamp859"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-28 19:06:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30859,"name":"M Camp","email":"mcamp859@hotmail.com","username":"mcamp859"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-28 17:54:08.000000000","updated_by":{"_account_id":33339,"name":"Kevin Putnam","email":"kevin.putnam@intel.com","username":"ikevinputnam"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2021-04-28 17:54:08.000000000","updated_by":{"_account_id":33339,"name":"Kevin Putnam","email":"kevin.putnam@intel.com","username":"ikevinputnam"},"reviewer":{"_account_id":30859,"name":"M Camp","email":"mcamp859@hotmail.com","username":"mcamp859"},"state":"REVIEWER"},{"updated":"2021-04-28 18:05:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"540b63063442ab108b719251403aec47166dc0e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33339,"name":"Kevin Putnam","email":"kevin.putnam@intel.com","username":"ikevinputnam"},"date":"2021-04-28 17:52:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46360f229a62a561b7186470fa8d71852947de00","author":{"_account_id":33339,"name":"Kevin Putnam","email":"kevin.putnam@intel.com","username":"ikevinputnam"},"date":"2021-04-28 17:54:08.000000000","message":"Patch Set 1:\n\n1st step towards promoting the release branch, hopefully.","accounts_in_message":[],"_revision_number":1},{"id":"3756e84b67334935e4fdc01000ee9324bf840407","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 18:05:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ba4af31b185a4f8eab0eb5e014b06c37 : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":1},{"id":"382dc6a8a14502110fd9965c8e1e88c19dd2caf1","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-04-28 18:17:07.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)\n\nHappy to approve once Mary confirms this is okay.\n\nTo clarify, the starlingx/docs repository\u0027s jobs were originally patterned after those for openstack/openstack-manuals which has only a master branch and keeps release-specific content in subtrees of the same branch. Since starlingx/docs is being branched, the former restriction to publish only from master was at odds with how its maintainers were working.\n\nI can\u0027t say for sure that this is going to do exactly what you want without spending a lot more time digging in the job definitions, but what I expect will happen is that the build for the r/stx.5.0 branch will start to get published at https://docs.starlingx.io/r/stx.5.0/ after this merges and the next change from that branch of starlingx/docs merges to trigger the promote-stx-tox-docs-site job for it. If that\u0027s not what happens, we\u0027ll have a bit more data and can adjust the approach to suit.","accounts_in_message":[],"_revision_number":1},{"id":"496028dddb1a4fb96eb7212715294270abdced1f","author":{"_account_id":30859,"name":"M Camp","email":"mcamp859@hotmail.com","username":"mcamp859"},"date":"2021-04-28 18:45:19.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nPlease update description and comment for history/tracking.","accounts_in_message":[],"_revision_number":1},{"id":"15d53c571e248981c68cf8ed566e9be1e1f4568a","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-04-28 18:48:28.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"9914042ac5f5502bfcc7dcb9ae7dba3bf386ea79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33339,"name":"Kevin Putnam","email":"kevin.putnam@intel.com","username":"ikevinputnam"},"date":"2021-04-28 18:56:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a354a7aee9ab6e0f56362b120ef4107584fe8ea7","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-04-28 18:59:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8527001779c067d85f031d52cdd0024376d77000","author":{"_account_id":30859,"name":"M Camp","email":"mcamp859@hotmail.com","username":"mcamp859"},"date":"2021-04-28 19:00:56.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"b5cd319e09cdc950c9da03a049ef144255446dd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 19:03:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/cfcfa2c6d7ab4bf4af8c1237109a0064 : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":2},{"id":"aff2d3c8808e724c03515b086af5b7c1d7e9a06e","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-04-28 19:06:02.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"71631263c01a44cc11084a3a558788a667fc963b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 19:06:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"933b9c8144c853deb6e29f2bf57e859a0debc3aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 19:21:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/70efb4e3a7464415b55415fe18a9e072 : SUCCESS in 10m 12s","accounts_in_message":[],"_revision_number":2},{"id":"175d810d3cce53e4616ed02c8d7a0deeb6dc93df","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 19:21:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6bd9d39d4d27d6c745c2c40231ddb8b3666db921","revisions":{"8a855db391bdbd44fa6c27468039710ca211ce38":{"kind":"REWORK","_number":1,"created":"2021-04-28 17:52:22.000000000","uploader":{"_account_id":33339,"name":"Kevin Putnam","email":"kevin.putnam@intel.com","username":"ikevinputnam"},"ref":"refs/changes/62/788562/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/62/788562/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/788562/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/788562/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/788562/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/62/788562/1"}}},"commit":{"parents":[{"commit":"6572d214905c751bed6f294c8dde1090ce40260d","subject":"Limit grafana for zk to zk04-06","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6572d214905c751bed6f294c8dde1090ce40260d"}]}],"author":{"name":"Kevin Putnam","email":"kevin.putnam@intel.com","date":"2021-04-28 17:51:07.000000000","tz":-420},"committer":{"name":"Kevin Putnam","email":"kevin.putnam@intel.com","date":"2021-04-28 17:51:07.000000000","tz":-420},"subject":"Adds r/stx.5.0 to the list of branches to promote for starlingx docs.","message":"Adds r/stx.5.0 to the list of branches to promote for starlingx docs.\n\nSigned-off-by: Kevin Putnam \u003ckevin.putnam@intel.com\u003e\nChange-Id: Ic1722c158d0f904cc7eb10bd09670725f3037dad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8a855db391bdbd44fa6c27468039710ca211ce38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8a855db391bdbd44fa6c27468039710ca211ce38"}]},"branch":"refs/heads/master"},"6bd9d39d4d27d6c745c2c40231ddb8b3666db921":{"kind":"REWORK","_number":2,"created":"2021-04-28 18:56:23.000000000","uploader":{"_account_id":33339,"name":"Kevin Putnam","email":"kevin.putnam@intel.com","username":"ikevinputnam"},"ref":"refs/changes/62/788562/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/62/788562/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/788562/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/788562/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/788562/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/62/788562/2"}}},"commit":{"parents":[{"commit":"6572d214905c751bed6f294c8dde1090ce40260d","subject":"Limit grafana for zk to zk04-06","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6572d214905c751bed6f294c8dde1090ce40260d"}]}],"author":{"name":"Kevin Putnam","email":"kevin.putnam@intel.com","date":"2021-04-28 17:51:07.000000000","tz":-420},"committer":{"name":"Kevin Putnam","email":"kevin.putnam@intel.com","date":"2021-04-28 18:54:19.000000000","tz":-420},"subject":"Adds r/stx.5.0 to the list of branches to promote for starlingx docs.","message":"Adds r/stx.5.0 to the list of branches to promote for starlingx docs.\n\nSigned-off-by: Kevin Putnam \u003ckevin.putnam@intel.com\u003e\nChange-Id: Ic1722c158d0f904cc7eb10bd09670725f3037dad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6bd9d39d4d27d6c745c2c40231ddb8b3666db921"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6bd9d39d4d27d6c745c2c40231ddb8b3666db921"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
