)]}'
{"id":"openstack%2Fproject-config~789053","triplet_id":"openstack%2Fproject-config~master~I58da8e838103223c54112229964e91d85f417cb4","project":"openstack/project-config","branch":"master","topic":"openstack-releases","hashtags":[],"change_id":"I58da8e838103223c54112229964e91d85f417cb4","subject":"Split vexxhost/base-jobs out to its own Gerrit ACL","status":"MERGED","created":"2021-05-01 13:17:17.000000000","updated":"2021-05-03 17:01:33.000000000","submitted":"2021-05-03 16:57:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"789053-openstack-releases","meta_rev_id":"ba23b1c062d0288a876c43e358f8b81f47a5dba1","_number":789053,"virtual_id_number":789053,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-03 16:57:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2021-05-03 16:01:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2021-05-01 17:51:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2021-05-01 21:40:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2021-05-03 16:51:55.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-03 16:51:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-01 13:23:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-01 17:51:09.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-05-01 21:40:03.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2021-05-03 16:01:39.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"58de4fa2a187c3545c76bf82c59b021a139b3b9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-01 13:17:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b234d639e9e4db0323691d0259f92a4de8f9487a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-01 13:23:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/03fc41187a3e4470b3553106183f4abe : SUCCESS in 5m 10s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/354d7f0db6d24745a2097e0906430f3b : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":1},{"id":"8010639104b058d81d20bec1ca9f714d96076151","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-05-01 17:51:09.000000000","message":"Patch Set 1: Code-Review+1\n\nSeems logical to split out the vexxhost gerrit ACLs.\nbase-jobs.config is identical with opendev/project-config.config, which was originally used to vexxhost/base-jobs.\nSo the patch looks good to me.","accounts_in_message":[],"_revision_number":1},{"id":"348921142e515b3d341431b4124d0ec2f794b4ea","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-05-01 21:40:03.000000000","message":"Patch Set 1: Code-Review+1\n\nIt makes sense to avoid corss-namespace reference in gerrit ACLs. The contents of the new config matches the previous one (i.e. opendev/project-config.config), so LGTM.\n\nOn the other hand, gerrit has no namespace for group naming. Perhaps it is better to use a namespace name as a prefix for a gerrit group for a new group. I don\u0027t think we need to adjust the existing groups.","accounts_in_message":[],"_revision_number":1},{"id":"224d4c13dc0ff82208b7d99f9a2bffda6873e401","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-01 22:00:40.000000000","message":"Patch Set 1:\n\nYes, I\u0027ve been pondering the namespace challenge for groups over the past year or two. I think we need to come up with a good scheme for them eventually, as we\u0027re bound to run into collisions at some point.","accounts_in_message":[],"_revision_number":1},{"id":"6048130c11cf028ac41ccdd40533fb615592b3ad","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-05-03 16:01:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b7f46493fcaf19f5c503f43c0e58c511446054d2","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-03 16:51:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"582279562f992559da24af5c4c8d05066af27709","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 16:51:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a957b4754fcf00392a9f51aaa06e09fd10c67efd","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-03 16:51:55.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"edc97cd0f93ab1b676bf9ccbeba764280856bfe6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 16:57:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d69ca20ec50a4e44bfabd997f15c6045 : SUCCESS in 5m 08s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/45692dceaaf54662a5c45202b1acc038 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"6ddef6c94da59630fc0549276c09cefbd50acdb8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 16:57:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ba23b1c062d0288a876c43e358f8b81f47a5dba1","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 17:01:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/4fa4cfda592d41ca9b3392acd9d937bc : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e1bf62ee90357273394284ba941aa48953f329a2","revisions":{"e1bf62ee90357273394284ba941aa48953f329a2":{"kind":"REWORK","_number":1,"created":"2021-05-01 13:17:17.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/53/789053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/53/789053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/53/789053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/53/789053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/53/789053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/53/789053/1"}}},"commit":{"parents":[{"commit":"a8820df4ecfd6792adf12ce1282d8ab21de9420d","subject":"Merge \"Move projects under meta-config acl (16)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a8820df4ecfd6792adf12ce1282d8ab21de9420d"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-01 13:13:19.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-01 13:16:50.000000000","tz":0},"subject":"Split vexxhost/base-jobs out to its own Gerrit ACL","message":"Split vexxhost/base-jobs out to its own Gerrit ACL\n\nThe vexxhost/base-jobs project started out sharing the Gerrit\nconfiguration for opendev/project-config, but cross-namespace ACLs\ncreate a bit of ambiguity. Give this repository a separate config in\nits own namespace.\n\nChange-Id: I58da8e838103223c54112229964e91d85f417cb4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e1bf62ee90357273394284ba941aa48953f329a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e1bf62ee90357273394284ba941aa48953f329a2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
