)]}'
{"id":"openstack%2Fproject-config~789977","triplet_id":"openstack%2Fproject-config~master~I5dc4f42629661da984cc197a11e3383cddc4ff54","project":"openstack/project-config","branch":"master","topic":"remove-arm64-xenial","hashtags":[],"change_id":"I5dc4f42629661da984cc197a11e3383cddc4ff54","subject":"Remove Xenial ARM64 nodes","status":"MERGED","created":"2021-05-06 02:17:25.000000000","updated":"2021-05-06 03:20:24.000000000","submitted":"2021-05-06 03:06:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"789977-remove-arm64-xenial","meta_rev_id":"b46972fedff481d01a4f878f3f608e66bb52ec20","_number":789977,"virtual_id_number":789977,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-06 03:06:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":2,"date":"2021-05-06 02:22:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-06 02:46:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-06 02:22:23.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2021-05-06 02:23:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0d9350d59cccb9abb5133b6f001f0bcb973bfb8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-05-06 02:17:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12be584230eaeabc3332fef7acad71b05b34abb8","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-06 02:22:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5a459efd3d3b000bde5379eb44c01f288101d2d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 02:23:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e41557f3481043fe9a9e8b41ad4c54dc : SUCCESS in 5m 31s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/bacd3c6240564dd6b0ba90ac2a13e6c0 : SUCCESS in 3m 42s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/33fd1375e0b048e59db7bf928f178084 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":1},{"id":"8ebd60b0f33132dc54b0650577a32e7550e48bba","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-05-06 02:46:07.000000000","message":"Patch Set 1: Workflow+1\n\ni will monitor this and make sure nb03 stops it\u0027s failing building loop","accounts_in_message":[],"_revision_number":1},{"id":"3741c684ec544bd6342240e6d8228462c8c2e621","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 02:46:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8d5f650452e2a2e541c225e3073c6917ffedd2ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 03:06:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2cb4b9aaf986421092160bc3f280ac2a : SUCCESS in 5m 14s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/dc62f0b026884387b92821172d94e440 : SUCCESS in 4m 06s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/ecf9fe16af1b430783c9b14e4d58ff5f : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"d396352cf9f34317277ec4a5b18d8d5eea8ccecd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 03:06:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b46972fedff481d01a4f878f3f608e66bb52ec20","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 03:20:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/e5588c1e6a6d455a856344b79841c879 : SUCCESS in 6m 21s\n- infra-prod-service-grafana https://zuul.opendev.org/t/openstack/build/32437e7aeec2432b8ed429100fc95612 : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8a8c39996c3f2198388448575a18165ae2f99ff3","revisions":{"8a8c39996c3f2198388448575a18165ae2f99ff3":{"kind":"REWORK","_number":1,"created":"2021-05-06 02:17:25.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/77/789977/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/77/789977/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/789977/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/789977/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/789977/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/77/789977/1"}}},"commit":{"parents":[{"commit":"808d35046e4ebf029a43c1637cacc0742b2f5cbf","subject":"Merge \"Replace old Wallaby cycle signing key with Xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/808d35046e4ebf029a43c1637cacc0742b2f5cbf"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-05-06 02:15:12.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-05-06 02:16:45.000000000","tz":600},"subject":"Remove Xenial ARM64 nodes","message":"Remove Xenial ARM64 nodes\n\nWheel jobs were removed with\nIa44c384072c0482cfd11c642013fd51004f85c8b.  There are now no more job\nreferences for these nodes, so this now removes them from nodepool.\n\nChange-Id: I5dc4f42629661da984cc197a11e3383cddc4ff54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8a8c39996c3f2198388448575a18165ae2f99ff3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8a8c39996c3f2198388448575a18165ae2f99ff3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
