)]}'
{"id":"openstack%2Fproject-config~791905","triplet_id":"openstack%2Fproject-config~master~Ia3c84e796dee97efea25609567125d05bd6fd299","project":"openstack/project-config","branch":"master","topic":"retire_panko","hashtags":[],"change_id":"Ia3c84e796dee97efea25609567125d05bd6fd299","subject":"Deprecate panko, python-pankoclient and puppet-panko","status":"MERGED","created":"2021-05-18 08:28:03.000000000","updated":"2021-05-27 16:32:53.000000000","submitted":"2021-05-27 16:20:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791905-retire_panko","meta_rev_id":"6b61a4ce2a03b927b29180957e517834f6f3f7d4","_number":791905,"virtual_id_number":791905,"owner":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-27 16:20:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2021-05-20 14:11:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2021-05-25 18:28:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-25 20:10:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2021-05-27 16:10:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2021-05-27 16:10:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-18 08:28:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-18 22:34:27.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"CC"},{"updated":"2021-05-19 15:23:15.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2021-05-25 18:28:40.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2021-05-25 20:10:42.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-05-27 16:10:19.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"89b6261a81c92e078d45ef8e70e22392abcbc2db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-05-18 08:28:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e67e3c8c91f20cad79d9a23c309ec90a044041a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 08:28:23.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  expected str @ data[\u0027templates\u0027][0]\n\nThe error appears in the following project stanza:\n\n  project:\n      name: openstack/panko\n      templates:\n        - official-openstack-repo-jobs:\n            - branches: ^(?!master).*$\n        - translation-jobs-master-only:\n            - branches: ^(?!master).*$\n            - publish-to-pypi:\n            - branches: ^(?!master).*$\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 4396, column 3","accounts_in_message":[],"_revision_number":1},{"id":"1c03ffc1e4c539f80bf747c1153433095a5de7ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-05-18 08:53:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"372d4a91a3d3e9a21b7f2ba7b3a85d774c0f56f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 08:53:57.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  expected str @ data[\u0027templates\u0027][0]\n\nThe error appears in the following project stanza:\n\n  project:\n      name: openstack/panko\n      templates:\n        - noop:\n            branches: master\n        - official-openstack-repo-jobs:\n            branches: ^stable/.*$\n        - translation-jobs-master-only:\n            branches: ^stable/.*$\n        - publish-to-pypi:\n            branches: ^stable/.*$\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 4396, column 3","accounts_in_message":[],"_revision_number":2},{"id":"acb8013558a44d8a006fc23ce6e15119c23d6cbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-05-18 09:05:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4046ef9a94443e52e935e30bb973e1a56516e7c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 09:16:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1ed692c92c9f4c00b24ebbb47e54b81c : SUCCESS in 6m 47s","accounts_in_message":[],"_revision_number":3},{"id":"837d75be3cae5e65810e085aa4b6f17859054d40","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-05-18 22:34:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"01879d6cdc45007537108a18e11b4b2fa8136679","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-05-19 06:57:04.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nThank you for your review. I\u0027ll drop it in the next change, which I\u0027ll send after this comment.","accounts_in_message":[],"_revision_number":3},{"id":"3bf46544536c78db5f912809066f6fac20947624","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-05-19 06:57:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d4d0552aae11c19436024f49b0169a5aae93a869","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 07:04:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a97ab0dfa0b546a1808c9993f7841c4f : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":4},{"id":"00800583eeb1689674c1f90178d82164f9fae66a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-05-19 15:23:15.000000000","message":"Patch Set 4: Code-Review-1\n\nAre you retiring or deprecating? See the difference in\nhttps://docs.openstack.org/project-team-guide/repository.html#deprecating-a-repository\nand https://docs.openstack.org/project-team-guide/repository.html#retiring-a-repository\n\nAnd then update the commit message or change for what you do.\n\n-1 since this says \"retire\" but the change is for deprecating.","accounts_in_message":[],"_revision_number":4},{"id":"c0404ebfc6a0d183135000fa509831200bfa002f","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-05-20 07:36:21.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4: Code-Review-1\n\u003e \n\u003e Are you retiring or deprecating? See the difference in\n\u003e https://docs.openstack.org/project-team-guide/repository.html#deprecating-a-repository\n\u003e and https://docs.openstack.org/project-team-guide/repository.html#retiring-a-repository\n\u003e \n\u003e And then update the commit message or change for what you do.\n\u003e \n\u003e -1 since this says \"retire\" but the change is for deprecating.\n\nThank you for your feedback. You are right, this is about deprecating before final removal.","accounts_in_message":[],"_revision_number":4},{"id":"0a7ccfb88e8d8aea5828044135089c447085d954","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-05-20 07:37:30.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1d72c2c591288f78cf027d4e44878c75cb492d5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 07:48:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3242ed6aa5e846939bb9211ada3b1466 : SUCCESS in 5m 47s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/473c66adb7364542b99c86e142d52b74 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":5},{"id":"d816a147827bc8c20d2c5cd85e220347a989a4d1","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-05-20 14:11:17.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"fed97aa10bf36e533c490989db9d463303781176","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-05-25 18:28:40.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"a5924458e7ab7491cfefe71a92c9e412d21e5858","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-05-25 20:10:42.000000000","message":"Patch Set 5: Code-Review+1\n\nlgtm. agree to deprecate it instead if retire as per stable policy.","accounts_in_message":[],"_revision_number":5},{"id":"c9fdb2bba3a5c1bfa54b1a6bf80911d16a67e649","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-27 16:10:19.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"ccd08e1555d25175aad8a930a317fabe4f7ed3ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 16:10:57.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"59630b54e06e064c01b6d37ff98167f169058ae5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 16:20:57.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1fec5f124c024200b7af9cc9a3681ba2 : SUCCESS in 6m 32s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/53c88fad5e454dc59a54cb0ecd5c44bb : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":5},{"id":"35393c9ef36672461de474e3dc74de150964b77e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 16:20:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"6b61a4ce2a03b927b29180957e517834f6f3f7d4","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 16:32:53.000000000","message":"Patch Set 5:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/60a295b7427948ba9f095e88208ff141 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d809369eb5a14dcfac704f908ef4e2c82ba638da","revisions":{"96a062040edfe84ef761f2917ac05c9216589ad7":{"kind":"REWORK","_number":1,"created":"2021-05-18 08:28:03.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/05/791905/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/05/791905/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/791905/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/791905/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/791905/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/05/791905/1"}}},"commit":{"parents":[{"commit":"53a8380728c35e8d45b4d70106881b19c74f1659","subject":"Merge \"Retire puppet-glare - Step 3: Remove Project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/53a8380728c35e8d45b4d70106881b19c74f1659"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-05-18 08:27:05.000000000","tz":120},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-05-18 08:27:05.000000000","tz":120},"subject":"Retire panko, python-pankoclient and puppet-panko","message":"Retire panko, python-pankoclient and puppet-panko\n\nin master.\n\nChange-Id: Ia3c84e796dee97efea25609567125d05bd6fd299\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/96a062040edfe84ef761f2917ac05c9216589ad7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/96a062040edfe84ef761f2917ac05c9216589ad7"}]},"branch":"refs/heads/master"},"a35bdd9dea1044517cc8d75a894ee65793eea48d":{"kind":"REWORK","_number":2,"created":"2021-05-18 08:53:37.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/05/791905/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/05/791905/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/791905/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/791905/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/791905/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/05/791905/2"}}},"commit":{"parents":[{"commit":"53a8380728c35e8d45b4d70106881b19c74f1659","subject":"Merge \"Retire puppet-glare - Step 3: Remove Project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/53a8380728c35e8d45b4d70106881b19c74f1659"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-05-18 08:27:05.000000000","tz":120},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-05-18 08:52:04.000000000","tz":120},"subject":"Retire panko, python-pankoclient and puppet-panko","message":"Retire panko, python-pankoclient and puppet-panko\n\nin master.\n\nChange-Id: Ia3c84e796dee97efea25609567125d05bd6fd299\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a35bdd9dea1044517cc8d75a894ee65793eea48d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a35bdd9dea1044517cc8d75a894ee65793eea48d"}]},"branch":"refs/heads/master"},"c969efc13fc1bd021720358e7e97e8a15f4fe58e":{"kind":"REWORK","_number":3,"created":"2021-05-18 09:05:55.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/05/791905/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/05/791905/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/791905/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/791905/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/791905/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/05/791905/3"}}},"commit":{"parents":[{"commit":"53a8380728c35e8d45b4d70106881b19c74f1659","subject":"Merge \"Retire puppet-glare - Step 3: Remove Project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/53a8380728c35e8d45b4d70106881b19c74f1659"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-05-18 08:27:05.000000000","tz":120},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-05-18 09:05:45.000000000","tz":120},"subject":"Retire panko, python-pankoclient and puppet-panko","message":"Retire panko, python-pankoclient and puppet-panko\n\nin master.\n\nChange-Id: Ia3c84e796dee97efea25609567125d05bd6fd299\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c969efc13fc1bd021720358e7e97e8a15f4fe58e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c969efc13fc1bd021720358e7e97e8a15f4fe58e"}]},"branch":"refs/heads/master"},"aed5325360e17279e58b4adf315ec873bcb4e7a4":{"kind":"REWORK","_number":4,"created":"2021-05-19 06:57:13.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/05/791905/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/05/791905/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/791905/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/791905/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/791905/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/05/791905/4"}}},"commit":{"parents":[{"commit":"53a8380728c35e8d45b4d70106881b19c74f1659","subject":"Merge \"Retire puppet-glare - Step 3: Remove Project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/53a8380728c35e8d45b4d70106881b19c74f1659"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-05-18 08:27:05.000000000","tz":120},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-05-19 06:56:09.000000000","tz":120},"subject":"Retire panko, python-pankoclient and puppet-panko","message":"Retire panko, python-pankoclient and puppet-panko\n\nin master.\n\nChange-Id: Ia3c84e796dee97efea25609567125d05bd6fd299\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/aed5325360e17279e58b4adf315ec873bcb4e7a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/aed5325360e17279e58b4adf315ec873bcb4e7a4"}]},"branch":"refs/heads/master"},"d809369eb5a14dcfac704f908ef4e2c82ba638da":{"kind":"REWORK","_number":5,"created":"2021-05-20 07:37:30.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/05/791905/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/05/791905/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/791905/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/791905/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/791905/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/05/791905/5"}}},"commit":{"parents":[{"commit":"53a8380728c35e8d45b4d70106881b19c74f1659","subject":"Merge \"Retire puppet-glare - Step 3: Remove Project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/53a8380728c35e8d45b4d70106881b19c74f1659"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-05-18 08:27:05.000000000","tz":120},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-05-20 07:36:13.000000000","tz":120},"subject":"Deprecate panko, python-pankoclient and puppet-panko","message":"Deprecate panko, python-pankoclient and puppet-panko\n\nin master.\n\nWhile telemetry followed the stable policy, it is not listed there.\nThere is no requirements syncing set up.\n\nChange-Id: Ia3c84e796dee97efea25609567125d05bd6fd299\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d809369eb5a14dcfac704f908ef4e2c82ba638da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d809369eb5a14dcfac704f908ef4e2c82ba638da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
