)]}'
{"id":"openstack%2Fproject-config~792843","triplet_id":"openstack%2Fproject-config~master~Ib3c43ef5ba22191d869629cd01d3800f3e235ea4","project":"openstack/project-config","branch":"master","topic":"oftc","hashtags":[],"change_id":"Ib3c43ef5ba22191d869629cd01d3800f3e235ea4","subject":"Switch the IRC access check to OFTC","status":"MERGED","created":"2021-05-24 18:54:37.000000000","updated":"2021-05-28 17:27:49.000000000","submitted":"2021-05-28 15:58:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"792843-oftc","meta_rev_id":"a4fe8077453119d5e9ef4ce96ed4fdf6f9878048","_number":792843,"virtual_id_number":792843,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-28 15:58:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2021-05-27 19:35:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2021-05-28 13:47:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-28 15:49:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-24 19:05:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-24 19:28:10.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"CC"},{"updated":"2021-05-26 14:52:17.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2021-05-27 19:35:44.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2021-05-28 13:47:32.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"cfdd373081f80ab1dfc48c1689a2ea8ed9547f4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-24 18:54:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4afcee155b23b19e7976856e92ebd3ee9af1e761","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-24 18:56:29.000000000","message":"Set Work In Progress\n\nNot to be merged unless we are moving to OFTC.","accounts_in_message":[],"_revision_number":1},{"id":"81f345b08d31b541fcb58097b3a558e5658b6708","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 19:05:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/e2f696c2c1e7409786474b851b948061 : SUCCESS in 4m 03s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/99d36191b4134cb7a5f35c5ff98883e2 : SUCCESS in 4m 53s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/4edbbd98adc74466b761e2babe3efc63 : SUCCESS in 4m 08s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/9a92cfb5ebd048adb5259d3874c78a7f : SUCCESS in 4m 16s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/257aec2e97d441d9b2b7a64274cc527c : SUCCESS in 5m 31s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/9dcc590038124057924cc2f36b88290f : SUCCESS in 4m 49s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/8c6c6703d5af461daf86b0f67f3767ed : FAILURE in 7m 03s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/ce3e7a3e30504401b6835a92f7ccab1f : SUCCESS in 7m 09s","accounts_in_message":[],"_revision_number":1},{"id":"72666bd9b1a4623e3b4a5c74595cefd6dd8b9d31","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-05-24 19:28:10.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8c48c9b76d0093fd170112328b8dfdb064346fd4","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-24 21:29:53.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"9833aa775c4e5400c3dfa7a946eda8b8f86a595b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-24 21:41:45.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"a56c26d039c54037c49445f99a3512b4bf6265a8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-24 21:44:40.000000000","message":"Topic changed from irc-logging to oftc","accounts_in_message":[],"_revision_number":2},{"id":"bba38064fdf0819ad7c8eba5dc572642aabd54f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 21:49:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/c983fa420d9a4126ad1cc561cd9798d3 : SUCCESS in 4m 21s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b20e37b514824eda8d01c92bde7c12a3 : SUCCESS in 5m 57s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/19b7280ae2c74ef29ff10643f8bbc2fd : SUCCESS in 4m 21s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/0015ba436292444da305c9cc28d4ee82 : SUCCESS in 4m 40s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/dcd7521628564e8b8767b9baa30114e8 : SUCCESS in 3m 53s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/ed6da5224b0148ffb56e78d8a8731b78 : SUCCESS in 4m 12s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/71c1538b769e49edb4bcd245b51f202b : SUCCESS in 6m 20s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/b75e746f41ca4b108634ef1f614a316d : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":2},{"id":"cb353cc318010353de5b5ccaa624c61a475be580","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-25 16:19:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d309c8cb35ebdef27bf47bbcb79be660f6502bc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 16:27:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/55f5e2868dea4e0389d3b71b85eb3206 : SUCCESS in 3m 54s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/80b6e2c1735c43f89b4c883655a41c5b : SUCCESS in 7m 12s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/ab7666f4046c40aeb1a1e3d4551088d4 : SUCCESS in 4m 12s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/7fd7aed6e6cb4f94b61f12e39fb62425 : SUCCESS in 7m 02s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/c07bae2e33ee4188b121274567c8caec : SUCCESS in 7m 11s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/70a19592add74aadaab53fd600e0c956 : SUCCESS in 7m 11s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/45aa6f097e054d38acb4f218661c4230 : SUCCESS in 6m 41s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/9ed1fa37182b46539c129526beba0524 : SUCCESS in 7m 00s","accounts_in_message":[],"_revision_number":3},{"id":"5ace91945110a5d24cd6b05896fb02ec866fef3f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-25 18:52:39.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"62f09d8eeef5a24553deb79cab9b1dbd9d96afaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 19:00:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/234d542af50744e0a9ec9db0d1d45d77 : SUCCESS in 3m 57s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3bcbafcceeba4ffea9ae44779bf1b104 : SUCCESS in 4m 56s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/333134b7071a44af9ef1088d841537d8 : SUCCESS in 4m 57s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/7541fdd46c6d411698d8265c8c4dcd6c : SUCCESS in 4m 14s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/dce0209be9124b25b12f1ae4182d4fb3 : SUCCESS in 4m 53s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/b15a4f9509d14d21ac8f30498ffb31ac : SUCCESS in 4m 08s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/8921b8219b8b4a7eb0fe84efb5ee739d : SUCCESS in 6m 15s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/c0ca10594bb34db4bd254155fedd7870 : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":4},{"id":"e6518f7012289e7f94f137f0bb27ec9e0fe12cf3","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-25 21:14:34.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"af54ac04649b260a45e11e100fdd4c4f09a0d5de","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-25 21:27:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"25e7598d3d517cb03d61619879d4e9e97638449d","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-25 23:57:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"81970a75fcf719d031c816df6a60d829fc463388","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2021-05-26 14:52:17.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"002fbd50d1549363f24220f55820d33b41db1593","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-26 16:56:25.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3c39e0265f3c2fab3c63dd6d28b7c4f7f72bc372","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 17:21:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/a35f909cf9f64fa4bbf4f2103cb74dd5 : SUCCESS in 5m 19s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a11a56b8679048dc8f1ce6cba430a4d8 : SUCCESS in 6m 38s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/6050b1c4b8224050bde69a04cef7325d : SUCCESS in 4m 35s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/7044cb27427947759a5ca91399e72de2 : SUCCESS in 5m 03s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/318304ed553c489a9a08e8665add98e4 : SUCCESS in 5m 32s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/479bc718039b4014a3f2a7483889669d : SUCCESS in 4m 28s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/51daad56f48940819a9cf9d002b89c76 : SUCCESS in 7m 59s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/149b68d193ea4aa3bf568f491d3305a3 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":5},{"id":"d13ba6cddb131eba47144d3d06705b6275a182ce","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-27 17:09:25.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"7dab11d9c97a924b65e613dd16cdc69638ac29c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 17:23:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/2af3a78534ef48c1b9525fa7bb24b864 : SUCCESS in 4m 15s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/12055d64b68f49b789580525ed6160b2 : SUCCESS in 5m 56s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/d19890429d2a456fb178c0d3cd8a2955 : SUCCESS in 4m 52s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/c15ab8fb445c4a98b879fc73c7661645 : SUCCESS in 5m 26s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/bce280e23c70448fbeb4a119483ab718 : SUCCESS in 4m 43s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/7bfc0ab5145a464093e70f6536af34ce : SUCCESS in 4m 45s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/c44aa07fcbbe4da4adfbb3bff38d7d80 : SUCCESS in 7m 00s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/48cc3de6d0bc4a7ebc2ed25e27ab0022 : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":6},{"id":"8152eeadefbc5c15ba2d02251bd6f7d577398268","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-05-27 19:35:44.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"b10cbfa61372f8e2c9e2a7879d0488a2b73a303b","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-28 12:32:20.000000000","message":"Patch Set 6:\n\nI\u0027ll un-WIP and approve this and Id11598fc42672359e1abef7b70cc23100b16ab12 at the same time, with deployment temporarily halted.","accounts_in_message":[],"_revision_number":6},{"id":"2418be7eb06c2e406179dcdb04a9f06267fae6eb","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-05-28 13:47:32.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"7fe62f8f02f1f8352a120f8baae63835974f06c8","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-28 15:46:50.000000000","message":"Set Ready For Review\n\nAnsible is temporarily disabled for this.","accounts_in_message":[],"_revision_number":6},{"id":"9bfc025ae94349f6d5dca4fe68793a7a2e288949","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-28 15:49:36.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"9912869f6e6dc4f448a0729bd0225edc637b2ded","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 15:50:19.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"8cdd00e310ea36e973322d58154dfd0719f7bf75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 15:58:11.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/ab91613b81b34a24b39967e8ea9957fe : SUCCESS in 4m 22s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d6f45e9657f647b7a1d38cd7819e4a84 : SUCCESS in 6m 05s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/f8e084fb1214496ca489188200cf382d : SUCCESS in 3m 53s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/21f6e7b93c8249028bd7b1273506a45a : SUCCESS in 4m 45s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/3aaf29f1a9b44d898717420e2485486b : SUCCESS in 4m 22s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/c119b524c78d4ec983f0f67b5331fe47 : SUCCESS in 4m 25s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/80a2291470914d3d87dfb0149589247d : SUCCESS in 6m 35s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/9979a8a105384ac9873b668680e8a2b3 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":6},{"id":"a868b9c211b6d065441c56bde3d281faec3b04a2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 15:58:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"a4fe8077453119d5e9ef4ce96ed4fdf6f9878048","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 17:27:49.000000000","message":"Patch Set 6:\n\nBuild succeeded (deploy pipeline).\n\n- promote-infra-index https://zuul.opendev.org/t/openstack/build/3e7dcb2aaf1e44718c1de9f5c78afa5b : SUCCESS in 53s\n- promote-openstack-specs-site https://zuul.opendev.org/t/openstack/build/bec2a94cd50647c8bda83641a0e1ac7a : SUCCESS in 1m 09s\n- infra-prod-run-accessbot https://zuul.opendev.org/t/openstack/build/ebd1941713ff45a599b14f42843298f8 : SUCCESS in 27m 49s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"e8580686431ce586fc04025085f865be405313be","revisions":{"9a9b42636369127a6e583d91a3aa356babc9fa95":{"kind":"REWORK","_number":1,"created":"2021-05-24 18:54:37.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/43/792843/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/43/792843/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/792843/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/792843/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/792843/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/43/792843/1"}}},"commit":{"parents":[{"commit":"476e315189ba56e4588b2be7c69a161f285f221c","subject":"Accessbot OFTC channel list stopgap","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/476e315189ba56e4588b2be7c69a161f285f221c"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-24 18:51:42.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-24 18:54:15.000000000","tz":0},"subject":"Switch the IRC access check to OFTC","message":"Switch the IRC access check to OFTC\n\nMake some adjustments to the IRC access check script so that it\nworks in the OFTC network now. Also update the channel config to\nreflect the new ACL paradigms there.\n\nChange-Id: Ib3c43ef5ba22191d869629cd01d3800f3e235ea4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9a9b42636369127a6e583d91a3aa356babc9fa95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9a9b42636369127a6e583d91a3aa356babc9fa95"}]},"branch":"refs/heads/master"},"a3973fcbb3ce798374f2bb8bcce4a1ab6c538561":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-05-24 21:41:45.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/43/792843/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/43/792843/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/792843/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/792843/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/792843/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/43/792843/2"}}},"commit":{"parents":[{"commit":"2180224020d740c968a745bb06448d6a24c94594","subject":"Accessbot OFTC channel list stopgap","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2180224020d740c968a745bb06448d6a24c94594"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-24 18:51:42.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-24 21:32:36.000000000","tz":0},"subject":"Switch the IRC access check to OFTC","message":"Switch the IRC access check to OFTC\n\nMake some adjustments to the IRC access check script so that it\nworks in the OFTC network now. Also update the channel config to\nreflect the new ACL paradigms there.\n\nChange-Id: Ib3c43ef5ba22191d869629cd01d3800f3e235ea4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a3973fcbb3ce798374f2bb8bcce4a1ab6c538561"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a3973fcbb3ce798374f2bb8bcce4a1ab6c538561"}]},"branch":"refs/heads/master"},"0a82870bb443042e878fde7e62fdbd7f5abc0a29":{"kind":"REWORK","_number":3,"created":"2021-05-25 16:19:14.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/43/792843/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/43/792843/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/792843/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/792843/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/792843/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/43/792843/3"}}},"commit":{"parents":[{"commit":"2180224020d740c968a745bb06448d6a24c94594","subject":"Accessbot OFTC channel list stopgap","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2180224020d740c968a745bb06448d6a24c94594"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-24 18:51:42.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-25 16:13:12.000000000","tz":0},"subject":"Switch the IRC access check to OFTC","message":"Switch the IRC access check to OFTC\n\nMake some adjustments to the IRC access check script so that it\nworks in the OFTC network now. Also update the channel config to\nreflect the new ACL paradigms there. Remove our volunteer operators\ntemporarily until we can confirm their nicks there. Also rip out the\nchannel forwarding for unregistered nicks, we can work on readding\nit there later if we determine it\u0027s necessary after all, but it will\nneed implementing differently anyway if so.\n\nChange-Id: Ib3c43ef5ba22191d869629cd01d3800f3e235ea4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0a82870bb443042e878fde7e62fdbd7f5abc0a29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0a82870bb443042e878fde7e62fdbd7f5abc0a29"}]},"branch":"refs/heads/master"},"ba004a959480f5807d8dbeda7f4b233dc32e6a05":{"kind":"REWORK","_number":4,"created":"2021-05-25 18:52:39.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/43/792843/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/43/792843/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/792843/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/792843/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/792843/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/43/792843/4"}}},"commit":{"parents":[{"commit":"2180224020d740c968a745bb06448d6a24c94594","subject":"Accessbot OFTC channel list stopgap","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2180224020d740c968a745bb06448d6a24c94594"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-24 18:51:42.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-25 17:56:51.000000000","tz":0},"subject":"Switch the IRC access check to OFTC","message":"Switch the IRC access check to OFTC\n\nMake some adjustments to the IRC access check script so that it\nworks in the OFTC network now. Also update the channel config to\nreflect the new ACL paradigms there. Remove our volunteer operators\ntemporarily until we can confirm their nicks there. Also rip out the\nchannel forwarding for unregistered nicks, we can work on readding\nit there later if we determine it\u0027s necessary after all, but it will\nneed implementing differently anyway if so.\n\nChange-Id: Ib3c43ef5ba22191d869629cd01d3800f3e235ea4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ba004a959480f5807d8dbeda7f4b233dc32e6a05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ba004a959480f5807d8dbeda7f4b233dc32e6a05"}]},"branch":"refs/heads/master"},"bab8218ba27630f198a5eb72769f1aeb189403c5":{"kind":"REWORK","_number":5,"created":"2021-05-26 16:56:25.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/43/792843/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/43/792843/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/792843/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/792843/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/792843/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/43/792843/5"}}},"commit":{"parents":[{"commit":"2180224020d740c968a745bb06448d6a24c94594","subject":"Accessbot OFTC channel list stopgap","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2180224020d740c968a745bb06448d6a24c94594"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-24 18:51:42.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-26 16:55:22.000000000","tz":0},"subject":"Switch the IRC access check to OFTC","message":"Switch the IRC access check to OFTC\n\nMake some adjustments to the IRC access check script so that it\nworks in the OFTC network now. Also update the channel config to\nreflect the new ACL paradigms there. Remove our volunteer operators\ntemporarily until we can confirm their nicks there. Also rip out the\nchannel forwarding for unregistered nicks, we can work on readding\nit there later if we determine it\u0027s necessary after all, but it will\nneed implementing differently anyway if so.\n\nChange-Id: Ib3c43ef5ba22191d869629cd01d3800f3e235ea4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bab8218ba27630f198a5eb72769f1aeb189403c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bab8218ba27630f198a5eb72769f1aeb189403c5"}]},"branch":"refs/heads/master"},"e8580686431ce586fc04025085f865be405313be":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-05-27 17:09:25.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/43/792843/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/43/792843/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/792843/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/792843/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/792843/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/43/792843/6"}}},"commit":{"parents":[{"commit":"9e35441f1ae77f8ac23ffca448bf0f1697c6aeb3","subject":"Clean up accessbot channels without services","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9e35441f1ae77f8ac23ffca448bf0f1697c6aeb3"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-24 18:51:42.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-27 17:09:04.000000000","tz":0},"subject":"Switch the IRC access check to OFTC","message":"Switch the IRC access check to OFTC\n\nMake some adjustments to the IRC access check script so that it\nworks in the OFTC network now. Also update the channel config to\nreflect the new ACL paradigms there. Remove our volunteer operators\ntemporarily until we can confirm their nicks there. Also rip out the\nchannel forwarding for unregistered nicks, we can work on readding\nit there later if we determine it\u0027s necessary after all, but it will\nneed implementing differently anyway if so.\n\nChange-Id: Ib3c43ef5ba22191d869629cd01d3800f3e235ea4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e8580686431ce586fc04025085f865be405313be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e8580686431ce586fc04025085f865be405313be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
