)]}'
{"id":"openstack%2Fproject-config~792857","triplet_id":"openstack%2Fproject-config~master~Icced679f6a3f2882e61c6c558357c1f0b3332dce","project":"openstack/project-config","branch":"master","topic":"oftc","hashtags":[],"change_id":"Icced679f6a3f2882e61c6c558357c1f0b3332dce","subject":"Revert \"Accessbot OFTC channel list stopgap\"","status":"MERGED","created":"2021-05-24 21:41:45.000000000","updated":"2021-06-15 15:56:27.000000000","submitted":"2021-06-15 15:42:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"792857-oftc","meta_rev_id":"9d294dc54db77f6e5d848b7ebe5a8dfba2bb13a4","_number":792857,"virtual_id_number":792857,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-15 15:42:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2021-06-15 06:23:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-15 15:31:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-24 21:50:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-15 06:23:00.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"d18f071cd77185ade380659d21a1bc9f6efa6467","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-24 21:41:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71f03dbc878622b2eae01e70a8b5b7a03966eee0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-24 21:44:42.000000000","message":"Topic changed from irc-logging to oftc","accounts_in_message":[],"_revision_number":1},{"id":"64b83701cf7d18f9f8a666cf70de722c50501b6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 21:50:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/063b5201066d46f1b07920ec5f616090 : SUCCESS in 5m 42s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/6f744d2899954cf3a83374599d42d97b : FAILURE in 6m 59s","accounts_in_message":[],"_revision_number":1},{"id":"37430da3ca632b2a69b6dd673d950e317cdca5dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-25 16:19:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"91bcd09b0e2ee3c7ec31a8ee904905860ffcebee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 16:28:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/62c82bf91c81420e8024010300e8b7cd : SUCCESS in 7m 34s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/8e04fbd58db14267841718c4e46686d1 : FAILURE in 6m 41s","accounts_in_message":[],"_revision_number":2},{"id":"0ad3adad7dd8ea358ee8a0eb0fdde8ef8a792245","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-25 18:52:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a35853b8df496634180234894cc50558857c1b34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 19:01:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/44f9891281354877a4ab41b959c01fde : SUCCESS in 6m 35s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/d5df9dd321ef401598c99268b042e986 : FAILURE in 6m 14s","accounts_in_message":[],"_revision_number":3},{"id":"536b59c10fb09563f10a67eecd33d91074953be3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-26 16:56:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ee602240e2aa149d8f651a4987fd74f9bd528edc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 17:09:11.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3b4aba571cef460888b2f6ebbec89d0a : SUCCESS in 6m 40s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/4ec836db12814c3ba541260c89001151 : FAILURE in 7m 08s","accounts_in_message":[],"_revision_number":4},{"id":"912b6eaa5c96150f97198dc5dbba382ec96626d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-27 17:09:25.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"9b70b4f7548f34d98a1a3e0e7bb75f716e386232","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 17:22:55.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7a99b48a6c49405a8160ebad9d0ce535 : SUCCESS in 6m 41s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/7bbaceedd4704c3584d05cce17bda9a7 : FAILURE in 6m 21s","accounts_in_message":[],"_revision_number":5},{"id":"18ed06b793ba51a77d048c249dd6939b97f824d7","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-29 13:46:27.000000000","message":"Patch Set 5:\n\nWe\u0027ll need some volunteer(s) to get up with the current channel admins and add our opendevaccess account before this can be reverted:\n\n    #edeploy (sbadia)\n    #openstack-sahara (SergeyLukjanov)\n    #puppet-openstack (xarses)\n    #rdo (apevec,jpena,spotz)","accounts_in_message":[],"_revision_number":5},{"id":"45f07db8c203d9ba27e0c67303c8cd7e758590e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-29 18:14:12.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"76ca54e4f1b23b582d0ff076d3f69b9f6c73fd8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-29 18:22:00.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/692ee3f4324d429784a26a271e2b8984 : SUCCESS in 5m 19s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/dddaf3ccac014b6a92f62049f4515923 : FAILURE in 6m 22s","accounts_in_message":[],"_revision_number":6},{"id":"4b2fb7d89d74db1bab59f7d74c74a9344eb89c7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-31 15:09:02.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"baa8be1290a29776305c5b2d07cdc304f3fc454d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-31 15:29:33.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"a6d5395c44a8ea08560ed06abe535232eda19824","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 15:37:57.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a459e4773ac44020b85f8891e43d253d : SUCCESS in 6m 32s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/373e085623b240fd8be7ffa8456ea5b0 : FAILURE in 7m 05s","accounts_in_message":[],"_revision_number":8},{"id":"2f760acd8c2e6caa8823ba3041f3a8554821b304","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-31 18:08:26.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"7e96db43b9a1c05fc2ed58acafcad5b4ff7fb2bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 18:15:53.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d9b5382936324de3bbc719f024c7e81a : SUCCESS in 6m 08s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/0fbc8ab8ed704bb28f8f7ad668560475 : FAILURE in 6m 32s","accounts_in_message":[],"_revision_number":9},{"id":"bc6991aca2872d3047ef7e66a2ef115d909fe0bd","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-06-14 19:44:56.000000000","message":"Patch Set 9:\n\nrecheck now that we should have access to the lone remaining channel","accounts_in_message":[],"_revision_number":9},{"id":"ff183163748c4928f05025b3addf785f56578513","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 19:52:16.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b4863d994e6142deae3046b66ae72268 : SUCCESS in 5m 19s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/ba90e2f42d5c4e6890107eadc719e7a9 : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":9},{"id":"bc1814975f167d1663c722c47da259c88a8ca728","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-06-15 06:23:00.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"83999dd88c4e8987d98231b092735b034a406b5d","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-06-15 15:31:29.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"f413f60beca38d67af09b1075a18c7e9c9620ec2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 15:31:47.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"0645a62115ee785448217c72461d060aac8bd77c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 15:42:18.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/06a0b86daca74fc180da06ca13313dc8 : SUCCESS in 7m 23s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/26f084d24625431798478e823397a132 : SUCCESS in 8m 20s","accounts_in_message":[],"_revision_number":9},{"id":"1a25d1c7bf037935b2664078bc4daea3c9c39ae0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 15:42:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"9d294dc54db77f6e5d848b7ebe5a8dfba2bb13a4","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 15:56:27.000000000","message":"Patch Set 9:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-run-accessbot https://zuul.opendev.org/t/openstack/build/f31d855fd81844cb9e2a6bc199dd7e9b : SUCCESS in 7m 35s\n- infra-prod-service-eavesdrop https://zuul.opendev.org/t/openstack/build/609a627bf26b4ec1be446c559678608f : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"1fc83e0c1f61db07aa9afba6f0edc342d90dc995","revisions":{"2a313caae584d9038d7e8eb55a266f5df33eb93d":{"kind":"REWORK","_number":1,"created":"2021-05-24 21:41:45.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/57/792857/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/792857/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/792857/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/792857/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/792857/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/792857/1"}}},"commit":{"parents":[{"commit":"a3973fcbb3ce798374f2bb8bcce4a1ab6c538561","subject":"Switch the IRC access check to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a3973fcbb3ce798374f2bb8bcce4a1ab6c538561"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-24 21:34:15.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-24 21:41:17.000000000","tz":0},"subject":"Revert \"Accessbot OFTC channel list stopgap\"","message":"Revert \"Accessbot OFTC channel list stopgap\"\n\nThis reverts commit 2180224020d740c968a745bb06448d6a24c94594.\n\nMerge once passing project-config-irc-access job indicates access is\nsufficient for these channels.\n\nChange-Id: Icced679f6a3f2882e61c6c558357c1f0b3332dce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2a313caae584d9038d7e8eb55a266f5df33eb93d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2a313caae584d9038d7e8eb55a266f5df33eb93d"}]},"branch":"refs/heads/master"},"fc4537053cd05fb4c54328e3f33b90009eb0a667":{"kind":"REWORK","_number":2,"created":"2021-05-25 16:19:14.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/57/792857/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/792857/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/792857/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/792857/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/792857/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/792857/2"}}},"commit":{"parents":[{"commit":"0a82870bb443042e878fde7e62fdbd7f5abc0a29","subject":"Switch the IRC access check to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0a82870bb443042e878fde7e62fdbd7f5abc0a29"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-24 21:34:15.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-25 16:15:11.000000000","tz":0},"subject":"Revert \"Accessbot OFTC channel list stopgap\"","message":"Revert \"Accessbot OFTC channel list stopgap\"\n\nThis reverts commit 2180224020d740c968a745bb06448d6a24c94594.\n\nMerge once passing project-config-irc-access job indicates access is\nsufficient for these channels.\n\nChange-Id: Icced679f6a3f2882e61c6c558357c1f0b3332dce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fc4537053cd05fb4c54328e3f33b90009eb0a667"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fc4537053cd05fb4c54328e3f33b90009eb0a667"}]},"branch":"refs/heads/master"},"2a05b0abbc386dea71f71c1ebdaf7f3edd4b809a":{"kind":"REWORK","_number":3,"created":"2021-05-25 18:52:39.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/57/792857/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/792857/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/792857/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/792857/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/792857/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/792857/3"}}},"commit":{"parents":[{"commit":"ba004a959480f5807d8dbeda7f4b233dc32e6a05","subject":"Switch the IRC access check to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ba004a959480f5807d8dbeda7f4b233dc32e6a05"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-24 21:34:15.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-25 17:56:52.000000000","tz":0},"subject":"Revert \"Accessbot OFTC channel list stopgap\"","message":"Revert \"Accessbot OFTC channel list stopgap\"\n\nThis reverts commit 2180224020d740c968a745bb06448d6a24c94594.\n\nMerge once passing project-config-irc-access job indicates access is\nsufficient for these channels.\n\nChange-Id: Icced679f6a3f2882e61c6c558357c1f0b3332dce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2a05b0abbc386dea71f71c1ebdaf7f3edd4b809a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2a05b0abbc386dea71f71c1ebdaf7f3edd4b809a"}]},"branch":"refs/heads/master"},"8ebac859d98dc5c29962b6489fa4763f7ac58c9d":{"kind":"REWORK","_number":4,"created":"2021-05-26 16:56:25.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/57/792857/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/792857/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/792857/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/792857/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/792857/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/792857/4"}}},"commit":{"parents":[{"commit":"bab8218ba27630f198a5eb72769f1aeb189403c5","subject":"Switch the IRC access check to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bab8218ba27630f198a5eb72769f1aeb189403c5"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-24 21:34:15.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-26 16:55:48.000000000","tz":0},"subject":"Revert \"Accessbot OFTC channel list stopgap\"","message":"Revert \"Accessbot OFTC channel list stopgap\"\n\nThis reverts commit 2180224020d740c968a745bb06448d6a24c94594.\n\nMerge once passing project-config-irc-access job indicates access is\nsufficient for these channels.\n\nChange-Id: Icced679f6a3f2882e61c6c558357c1f0b3332dce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8ebac859d98dc5c29962b6489fa4763f7ac58c9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8ebac859d98dc5c29962b6489fa4763f7ac58c9d"}]},"branch":"refs/heads/master"},"f2b67e7ea5346e7d7edc182ba7fdab96e87a6f28":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-05-27 17:09:25.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/57/792857/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/792857/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/792857/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/792857/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/792857/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/792857/5"}}},"commit":{"parents":[{"commit":"e8580686431ce586fc04025085f865be405313be","subject":"Switch the IRC access check to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e8580686431ce586fc04025085f865be405313be"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-24 21:34:15.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-27 17:09:04.000000000","tz":0},"subject":"Revert \"Accessbot OFTC channel list stopgap\"","message":"Revert \"Accessbot OFTC channel list stopgap\"\n\nThis reverts commit 2180224020d740c968a745bb06448d6a24c94594.\n\nMerge once passing project-config-irc-access job indicates access is\nsufficient for these channels.\n\nChange-Id: Icced679f6a3f2882e61c6c558357c1f0b3332dce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f2b67e7ea5346e7d7edc182ba7fdab96e87a6f28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f2b67e7ea5346e7d7edc182ba7fdab96e87a6f28"}]},"branch":"refs/heads/master"},"f9c9f9e3734d4c6d5dbf4cb6472b0361c28e2c4c":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2021-05-29 18:14:12.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/57/792857/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/792857/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/792857/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/792857/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/792857/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/792857/6"}}},"commit":{"parents":[{"commit":"e8580686431ce586fc04025085f865be405313be","subject":"Switch the IRC access check to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e8580686431ce586fc04025085f865be405313be"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-24 21:34:15.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-29 18:13:55.000000000","tz":0},"subject":"Revert \"Accessbot OFTC channel list stopgap\"","message":"Revert \"Accessbot OFTC channel list stopgap\"\n\nThis reverts commit 68c091e744c306b93d6bce1c095d29c58dbb0bf2.\n\nMerge once passing project-config-irc-access job indicates access is\nsufficient for these channels.\n\nChange-Id: Icced679f6a3f2882e61c6c558357c1f0b3332dce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f9c9f9e3734d4c6d5dbf4cb6472b0361c28e2c4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f9c9f9e3734d4c6d5dbf4cb6472b0361c28e2c4c"}]},"branch":"refs/heads/master"},"119384e93351e8d73206a9ae49615a784509a8a9":{"kind":"REWORK","_number":7,"created":"2021-05-31 15:09:02.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/57/792857/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/792857/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/792857/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/792857/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/792857/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/792857/7"}}},"commit":{"parents":[{"commit":"4b69049631772e95e7483660a6a1d6ed3ef8d812","subject":"Readd accessbot and gerritbot for #rdo on OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4b69049631772e95e7483660a6a1d6ed3ef8d812"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-24 21:34:15.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-31 15:08:16.000000000","tz":0},"subject":"Revert \"Accessbot OFTC channel list stopgap\"","message":"Revert \"Accessbot OFTC channel list stopgap\"\n\nThis reverts commit 68c091e744c306b93d6bce1c095d29c58dbb0bf2.\n\nMerge once passing project-config-irc-access job indicates access is\nsufficient for these channels.\n\nChange-Id: Icced679f6a3f2882e61c6c558357c1f0b3332dce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/119384e93351e8d73206a9ae49615a784509a8a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/119384e93351e8d73206a9ae49615a784509a8a9"}]},"branch":"refs/heads/master"},"fa7becc11c93ea2e2849c596a2615dd2bac21567":{"kind":"REWORK","_number":8,"created":"2021-05-31 15:29:33.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/57/792857/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/792857/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/792857/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/792857/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/792857/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/792857/8"}}},"commit":{"parents":[{"commit":"4b69049631772e95e7483660a6a1d6ed3ef8d812","subject":"Readd accessbot and gerritbot for #rdo on OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4b69049631772e95e7483660a6a1d6ed3ef8d812"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-24 21:34:15.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-31 15:25:39.000000000","tz":0},"subject":"Revert \"Accessbot OFTC channel list stopgap\"","message":"Revert \"Accessbot OFTC channel list stopgap\"\n\nThis reverts commit 68c091e744c306b93d6bce1c095d29c58dbb0bf2, except\nfor #rdo (fixed in a parent change) and #edeploy (configmed with\nEmilienM in #tripleo that the #edeploy channel is no longer needed\nso removed statusbot configuration for it with change\nI47613e4d88f358db5c51f9cb21c1eb1c2ebd32ed).\n\nMerge once passing project-config-irc-access job indicates access is\nsufficient for these channels.\n\nChange-Id: Icced679f6a3f2882e61c6c558357c1f0b3332dce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fa7becc11c93ea2e2849c596a2615dd2bac21567"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fa7becc11c93ea2e2849c596a2615dd2bac21567"}]},"branch":"refs/heads/master"},"1fc83e0c1f61db07aa9afba6f0edc342d90dc995":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":9,"created":"2021-05-31 18:08:26.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/57/792857/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/792857/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/792857/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/792857/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/792857/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/792857/9"}}},"commit":{"parents":[{"commit":"e09885a03b8c9e41de28999949461bfb82043f09","subject":"Readd accessbot/gerritbot for #puppet-openstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e09885a03b8c9e41de28999949461bfb82043f09"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-24 21:34:15.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-31 18:07:15.000000000","tz":0},"subject":"Revert \"Accessbot OFTC channel list stopgap\"","message":"Revert \"Accessbot OFTC channel list stopgap\"\n\nThis reverts commit 68c091e744c306b93d6bce1c095d29c58dbb0bf2, except\nfor #rdo and #puppet-openstack (fixed in parent changes) and\nchannel is no longer needed so removed statusbot configuration for\nit with change I47613e4d88f358db5c51f9cb21c1eb1c2ebd32ed).\n\nMerge once passing project-config-irc-access job indicates access is\nsufficient for these channels.\n\nChange-Id: Icced679f6a3f2882e61c6c558357c1f0b3332dce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1fc83e0c1f61db07aa9afba6f0edc342d90dc995"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1fc83e0c1f61db07aa9afba6f0edc342d90dc995"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
