)]}'
{"id":"openstack%2Fproject-config~793459","triplet_id":"openstack%2Fproject-config~master~I68a66f3a4aa32110ea2458c187d878a93ce9e272","project":"openstack/project-config","branch":"master","topic":"retire-gearman-plugin","hashtags":[],"change_id":"I68a66f3a4aa32110ea2458c187d878a93ce9e272","subject":"Remove x/gearman-plugin from infrastructure","status":"MERGED","created":"2021-05-27 15:30:02.000000000","updated":"2021-06-02 14:40:58.000000000","submitted":"2021-06-02 14:17:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"793459-retire-gearman-plugin","meta_rev_id":"5204d4c5db09aa9cfd848110678c51ef45115a4b","_number":793459,"virtual_id_number":793459,"owner":{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-02 14:17:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2021-05-28 09:00:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2021-06-02 12:24:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-27 16:31:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-02 14:07:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-27 15:45:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-27 16:31:48.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2021-05-28 09:00:35.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"b030ae4362fb1553fec966668531fbd975ea02a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},"date":"2021-05-27 15:30:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f817c1ef1761bc1fa3b78115ad419a343be08594","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 15:45:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e2d8df5d1a7b4b59b0f63be9840ec57b : SUCCESS in 10m 33s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/ce2bf4b1e2fc4981a93b90536fb04192 : SUCCESS in 7m 21s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/1648fc652cd94518a71584c757038792 : FAILURE in 10m 09s","accounts_in_message":[],"_revision_number":1},{"id":"af942cbee90cbcaf565ac29c51b20d1ee56f76a0","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-27 16:31:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2adf9a59320e5049c8ea05e3115f39e301080593","author":{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},"date":"2021-05-27 22:24:20.000000000","message":"Patch Set 1:\n\nBefore making the repository read-only in Gerrit, one would probably want to abandon the changes that are still open: https://review.opendev.org/q/project:x/gearman-plugin+is:open  mentioning the repo got retired in favor of  https://github.com/jenkinsci/gearman-plugin/","accounts_in_message":[],"_revision_number":1},{"id":"97de6dc2f76cc5eb16e736abbe00f57f2a3cc379","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-05-28 09:00:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0e49574df9b25468d1a6579ad5b7b2333b83223e","author":{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},"date":"2021-06-02 12:21:48.000000000","message":"Patch Set 1:\n\nrecheck after zuul config has been deployed","accounts_in_message":[],"_revision_number":1},{"id":"4f58b2fcdf9aaebc9fd67126113411371bb30aea","author":{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},"date":"2021-06-02 12:24:02.000000000","message":"Patch Set 1: Code-Review+1\n\nAll pending changes have been abandoned by Jeremy and there are none left open. https://review.opendev.org/q/project:x/gearman-plugin+is:open","accounts_in_message":[],"_revision_number":1},{"id":"4c773754b04592b82e7f67e10c94ae2e080fbdad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 12:31:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0d9653df96974c2bbc5185d6650184e4 : SUCCESS in 7m 39s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/6786f843eadb46b5bce8d3104e7bf111 : SUCCESS in 5m 51s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/5e7d8ce9877048eeb4c6d5af648a6022 : SUCCESS in 7m 09s","accounts_in_message":[],"_revision_number":1},{"id":"f1701c5b7cfa3c2acb9d563acff5c7664daa6c9c","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-06-02 14:07:30.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5c7f12a04ee58af6f88241edb5122cb14d13dfb6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 14:08:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1c5e463a2fd9794598ed80d1ae7c6ea2f0c7d6f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 14:17:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/18d81bc65d504192a0ebca18e4f7217c : SUCCESS in 7m 01s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f1508ff4707247968fce57359645600a : SUCCESS in 5m 24s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/2f276c1563d74b4cb5815a251842567a : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":1},{"id":"494f3727950f72585fb5ca1ad09c587ae175063d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 14:17:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5204d4c5db09aa9cfd848110678c51ef45115a4b","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 14:40:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/a54486fce9934ddea862072c365d6930 : SUCCESS in 3m 23s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/66dead97973c44c6b967210acd840d58 : SUCCESS in 7m 27s\n- infra-prod-service-eavesdrop https://zuul.opendev.org/t/openstack/build/4dd7ed3073904937b475538b9c3b9301 : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"96608ab270e37cb4f275b7f0a00814e7d824a99c","revisions":{"96608ab270e37cb4f275b7f0a00814e7d824a99c":{"kind":"REWORK","_number":1,"created":"2021-05-27 15:30:02.000000000","uploader":{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},"ref":"refs/changes/59/793459/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/59/793459/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/793459/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/793459/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/793459/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/59/793459/1"}}},"commit":{"parents":[{"commit":"4c10c23e827e9f6befaa83b653feb0eba7619100","subject":"Retire x/gearman-plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4c10c23e827e9f6befaa83b653feb0eba7619100"}]}],"author":{"name":"Antoine Musso","email":"hashar@free.fr","date":"2021-05-27 15:28:55.000000000","tz":120},"committer":{"name":"Antoine Musso","email":"hashar@free.fr","date":"2021-05-27 15:28:55.000000000","tz":120},"subject":"Remove x/gearman-plugin from infrastructure","message":"Remove x/gearman-plugin from infrastructure\n\nDepends-On: I0c84db2ad3fbb4d9f0eff793a0159c6ed3a8e25c\nDepends-On: Ib6010d7ce85a934501c50a53e9ac78dcf74bc403\nChange-Id: I68a66f3a4aa32110ea2458c187d878a93ce9e272\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/96608ab270e37cb4f275b7f0a00814e7d824a99c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/96608ab270e37cb4f275b7f0a00814e7d824a99c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
