)]}'
{"id":"openstack%2Fproject-config~794351","triplet_id":"openstack%2Fproject-config~master~I72e6da5e7f09dd89f9bf93acdf8c95263cf3bd69","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2021-10-20 22:19:15.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I72e6da5e7f09dd89f9bf93acdf8c95263cf3bd69","subject":"pti-python-tarball: Add option to install bindep","status":"ABANDONED","created":"2021-06-03 00:01:25.000000000","updated":"2021-10-20 22:19:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e209f7a0b0985ee8dffbd6f67dc0370e3cc24207","_number":794351,"virtual_id_number":794351,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-06-03 00:13:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-06-03 04:51:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-03 00:13:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-03 04:51:07.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2021-10-19 15:41:17.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"}],"messages":[{"id":"5e3e9b0b7e20619a06681be601269580eb81386a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-03 00:01:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2db211502c5d94a7d5c5743db10926bf65ccf13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 00:13:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5292a0d9252f4b449d09d4cde7ee9864 : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":1},{"id":"e6419ae91e58ebad9ddd793abcc2d1358820bd97","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-03 00:14:49.000000000","message":"Patch Set 1:\n\nSo I was hoping to use this to fix pyeclib\u0027s gate in https://review.opendev.org/c/openstack/pyeclib/+/794352/ but it looks like I didn\u0027t quite do it right...","accounts_in_message":[],"_revision_number":1},{"id":"4e315d8804ee3bf69fd5434921e908339eb43a59","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-06-03 04:51:07.000000000","message":"Patch Set 1: Code-Review-1\n\n\u003e Patch Set 1:\n\u003e \n\u003e So I was hoping to use this to fix pyeclib\u0027s gate in https://review.opendev.org/c/openstack/pyeclib/+/794352/ but it looks like I didn\u0027t quite do it right...\n\nBecause this is a privileged repo, the depends-on won\u0027t work. But why is a release process, that in theory should just be making a tarball, require bindep dependencies?","accounts_in_message":[],"_revision_number":1},{"id":"aa5c588219610c05ef3ba273e45e531fe063908a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-03 16:13:30.000000000","message":"Patch Set 1:\n\nIt\u0027s not just making a tarball, though -- it\u0027s also making wheels, which may need to compile and link against some C code.\n\nI suppose another option would be to make the wheel-building optional, and turn it off for pyeclib. I was thinking about doing that *regardless* for the publish job, but was inclined to allow the check job to build \u0026 check a wheel.","accounts_in_message":[],"_revision_number":1},{"id":"81005de7b514098a107a3bf882d2284658e0bd7b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-12 05:33:54.000000000","message":"Patch Set 1:\n\nFWIW, I tried making wheel-building optional in https://review.opendev.org/c/zuul/zuul-jobs/+/795419, but it seems to be having some CI trouble on xenial. (At minimum; I have no idea what\u0027s up with the Software Factory CI jobs.)","accounts_in_message":[],"_revision_number":1},{"id":"822f2515852a1bac59647b0610882a78016f2e40","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-10-19 15:41:17.000000000","message":"Patch Set 1:\n\nEven if you got it to work, the next hurdle would be that you can\u0027t upload platform-specific Linux wheels to PyPI, you\u0027d need to build them in a special \"manylinux\" meta-platform representative environment instead: https://www.python.org/dev/peps/pep-0513/","accounts_in_message":[],"_revision_number":1},{"id":"e209f7a0b0985ee8dffbd6f67dc0370e3cc24207","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-10-20 22:19:15.000000000","message":"Abandoned\n\nDropping in favor of just making wheel-building optional in https://review.opendev.org/c/zuul/zuul-jobs/+/795419","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7dcbfe257d042be8883cf206c28014f5ca0d55e0","revisions":{"7dcbfe257d042be8883cf206c28014f5ca0d55e0":{"kind":"REWORK","_number":1,"created":"2021-06-03 00:01:25.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/51/794351/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/51/794351/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/51/794351/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/51/794351/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/51/794351/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/51/794351/1"}}},"commit":{"parents":[{"commit":"4bca3fe769646f723987562cc565126272570316","subject":"Merge \"Revert \"Disable limestone due to mirror issues\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4bca3fe769646f723987562cc565126272570316"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-06-02 23:45:42.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-06-02 23:45:42.000000000","tz":-420},"subject":"pti-python-tarball: Add option to install bindep","message":"pti-python-tarball: Add option to install bindep\n\nDefault to false to preserve existing behavior.\n\nChange-Id: I72e6da5e7f09dd89f9bf93acdf8c95263cf3bd69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7dcbfe257d042be8883cf206c28014f5ca0d55e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7dcbfe257d042be8883cf206c28014f5ca0d55e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
