)]}'
{"id":"openstack%2Fproject-config~795599","triplet_id":"openstack%2Fproject-config~master~Ic7aee1df77510e1421d9b42fcd3f950da987a1c4","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"Ic7aee1df77510e1421d9b42fcd3f950da987a1c4","subject":"Removing openstack-state-management from the bots","status":"MERGED","created":"2021-06-09 17:04:53.000000000","updated":"2021-06-15 08:41:30.000000000","submitted":"2021-06-15 06:37:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795599","meta_rev_id":"a11408118cb68982cc46749f14035dfa357d135c","_number":795599,"virtual_id_number":795599,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-15 06:37:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2021-06-09 17:20:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2021-06-15 06:27:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2021-06-15 06:27:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-09 17:20:31.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2021-06-09 17:24:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-15 06:27:15.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"da8d4e887cd24e90ffaf06912ccfe1da57947701","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-09 17:04:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a89357585a603b841d1fb0254d4eab4bb80e2cd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-09 17:11:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5f28f13f853479f433778e096f86bcf45dd6024a","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-06-09 17:20:31.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"540ec663b20a0323dd5e7e7154b2934dba772da8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 17:24:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/830fb1e864004962abe959209a921357 : SUCCESS in 5m 42s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/b8d3ffae50354b0bad5a2104bb1514a4 : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":2},{"id":"fd7ee9d01d5585d6928d1ad57eab77664dc353fe","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-06-15 06:27:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5d727099a0e829e416571032b467e0f4cd85573b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 06:27:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"67e63de4aced19bed8657413b47b75e4f94c19e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 06:37:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b5028d2f3aa9498eafe42faf36c4f64b : SUCCESS in 7m 14s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/8244eff6de29449a9f8beef9f315e16c : SUCCESS in 8m 32s","accounts_in_message":[],"_revision_number":2},{"id":"e74784509aae2e26fc1e5dd2b76b8ef96aa6e7d1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 06:37:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a11408118cb68982cc46749f14035dfa357d135c","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 08:41:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-run-accessbot https://zuul.opendev.org/t/openstack/build/297fad0133b242d694b5bfeff75c67e7 : SUCCESS in 8m 13s\n- infra-prod-service-eavesdrop https://zuul.opendev.org/t/openstack/build/609262b0e81b444fa3bd19c0eb903eae : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"748b73ca95b937757c7cc60fd054c03ef6457896","revisions":{"43878a24a10e834b8dff184ca268a3f9b6cc67ec":{"kind":"REWORK","_number":1,"created":"2021-06-09 17:04:53.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/99/795599/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/99/795599/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/99/795599/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/99/795599/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/99/795599/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/99/795599/1"}}},"commit":{"parents":[{"commit":"328a90d4ae4bd5c1bb6a2f1604abd0cc69320a18","subject":"Merge \"Move devstack-gate jobs list to in-tree\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/328a90d4ae4bd5c1bb6a2f1604abd0cc69320a18"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-06-09 17:02:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-06-09 17:04:51.000000000","tz":0},"subject":"Removing openstack-state-management from gerritbot","message":"Removing openstack-state-management from gerritbot\n\nWe are retiring the #openstack-state-management channel in favor\nof consolidating the discussion to the #openstack-oslo channel[1].\nThis patch removes openstack-state-management from gerritbot.\n\nNote: This patch does not remove openstack-state-management from accessbot\nto retain control of the channel until we completely remove it from OFTC.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2021-June/022929.html\n\nChange-Id: Ic7aee1df77510e1421d9b42fcd3f950da987a1c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/43878a24a10e834b8dff184ca268a3f9b6cc67ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/43878a24a10e834b8dff184ca268a3f9b6cc67ec"}]},"branch":"refs/heads/master"},"748b73ca95b937757c7cc60fd054c03ef6457896":{"kind":"REWORK","_number":2,"created":"2021-06-09 17:11:37.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/99/795599/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/99/795599/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/99/795599/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/99/795599/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/99/795599/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/99/795599/2"}}},"commit":{"parents":[{"commit":"328a90d4ae4bd5c1bb6a2f1604abd0cc69320a18","subject":"Merge \"Move devstack-gate jobs list to in-tree\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/328a90d4ae4bd5c1bb6a2f1604abd0cc69320a18"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-06-09 17:02:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-06-09 17:11:02.000000000","tz":0},"subject":"Removing openstack-state-management from the bots","message":"Removing openstack-state-management from the bots\n\nWe are retiring the #openstack-state-management channel in favor\nof consolidating the discussion to the #openstack-oslo channel[1].\nThis patch removes openstack-state-management from gerritbot and\naccessbot.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2021-June/022929.html\n\nChange-Id: Ic7aee1df77510e1421d9b42fcd3f950da987a1c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/748b73ca95b937757c7cc60fd054c03ef6457896"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/748b73ca95b937757c7cc60fd054c03ef6457896"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
