)]}'
{"id":"openstack%2Fproject-config~795824","triplet_id":"openstack%2Fproject-config~master~Iafae4acf9fe561277f91ac9c7a2e4e60b9d58c84","project":"openstack/project-config","branch":"master","topic":"etcd3gw-release-config","hashtags":[],"change_id":"Iafae4acf9fe561277f91ac9c7a2e4e60b9d58c84","subject":"Adding missing zuul for etcd3gw","status":"MERGED","created":"2021-06-10 14:48:29.000000000","updated":"2021-06-29 00:43:21.000000000","submitted":"2021-06-29 00:43:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795824-etcd3gw-release-config","meta_rev_id":"c139f1e72c1e6dc37c0ab38329229421775a8eba","_number":795824,"virtual_id_number":795824,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-29 00:43:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2021-06-15 06:23:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-29 00:28:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-29 00:28:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-10 14:58:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-15 06:23:21.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2021-06-15 15:02:49.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"e8eec8915c9ee8aacbcb5a6f9d484264d78ca937","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-10 14:48:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff1cd123c24735d339244757586c1fb07ca8eaea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 14:58:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/212d03b4c7524e2b80f66bb180150c95 : FAILURE in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"145f54c51e4f91d9d60ef8c76d3164bf549eb3ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-14 14:26:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"21c9f5c54014f90f3ce61fa6bbd1f719e7be7639","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 14:34:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9414b251863d4e148a2dfd9a392ec24c : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":2},{"id":"5c631802c84b62df37622256b5b5002817e078cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-14 15:01:01.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"59a71adb112b26c4d8e5445f6c883165f2e10f7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 15:16:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/fb3c14fc80eb499e9b6f9819fe6a17f8 : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":3},{"id":"044b447accf9732dff56599bee12db3765ce447e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-06-15 06:23:21.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f5c99837abe38bd6dd6235bad5d25fb8a4ebe133","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-06-29 00:28:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6a823d57978c1e45f517950da1ac93d8ee49e1c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 00:29:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4408f1a3ef6d8b56709f0ccc958bdb46087d3d66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 00:43:17.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e9612845ec9341dc8c498c323322bbf8 : SUCCESS in 7m 08s","accounts_in_message":[],"_revision_number":3},{"id":"c139f1e72c1e6dc37c0ab38329229421775a8eba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 00:43:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"34b136c75dd2a2f1357f153cfbd68d16f2dcd91e","revisions":{"af801a918a0351dc1618b7cb504167c4e8756e3c":{"kind":"REWORK","_number":1,"created":"2021-06-10 14:48:29.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/24/795824/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/24/795824/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/795824/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/795824/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/795824/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/24/795824/1"}}},"commit":{"parents":[{"commit":"328a90d4ae4bd5c1bb6a2f1604abd0cc69320a18","subject":"Merge \"Move devstack-gate jobs list to in-tree\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/328a90d4ae4bd5c1bb6a2f1604abd0cc69320a18"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-10 14:44:48.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-10 14:44:48.000000000","tz":120},"subject":"Adding missing zuul for etcd3gw","message":"Adding missing zuul for etcd3gw\n\netcd3gw have been added to the oslo scope few months ago however it was\nnever released and some elements are missing to allow us to release it.\n\nhttps://opendev.org/openstack/governance/commit/e90a44648cb2224f9ca5efef6f10ffea963fe6d1\n\nChange-Id: Iafae4acf9fe561277f91ac9c7a2e4e60b9d58c84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/af801a918a0351dc1618b7cb504167c4e8756e3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/af801a918a0351dc1618b7cb504167c4e8756e3c"}]},"branch":"refs/heads/master"},"3c96aaab0e1d7934ddee79b6a7c5ab349934f1bf":{"kind":"REWORK","_number":2,"created":"2021-06-14 14:26:01.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/24/795824/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/24/795824/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/795824/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/795824/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/795824/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/24/795824/2"}}},"commit":{"parents":[{"commit":"328a90d4ae4bd5c1bb6a2f1604abd0cc69320a18","subject":"Merge \"Move devstack-gate jobs list to in-tree\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/328a90d4ae4bd5c1bb6a2f1604abd0cc69320a18"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-10 14:44:48.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-14 14:25:48.000000000","tz":120},"subject":"Adding missing zuul for etcd3gw","message":"Adding missing zuul for etcd3gw\n\netcd3gw have been added to the oslo scope few months ago however it was\nnever released and some elements are missing to allow us to release it.\n\nhttps://opendev.org/openstack/governance/commit/e90a44648cb2224f9ca5efef6f10ffea963fe6d1\n\nChange-Id: Iafae4acf9fe561277f91ac9c7a2e4e60b9d58c84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3c96aaab0e1d7934ddee79b6a7c5ab349934f1bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3c96aaab0e1d7934ddee79b6a7c5ab349934f1bf"}]},"branch":"refs/heads/master"},"34b136c75dd2a2f1357f153cfbd68d16f2dcd91e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-06-14 15:01:01.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/24/795824/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/24/795824/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/795824/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/795824/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/795824/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/24/795824/3"}}},"commit":{"parents":[{"commit":"328a90d4ae4bd5c1bb6a2f1604abd0cc69320a18","subject":"Merge \"Move devstack-gate jobs list to in-tree\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/328a90d4ae4bd5c1bb6a2f1604abd0cc69320a18"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-10 14:44:48.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-14 15:00:19.000000000","tz":120},"subject":"Adding missing zuul for etcd3gw","message":"Adding missing zuul for etcd3gw\n\netcd3gw has been added to the oslo scope few months ago however it was\nnever released and some elements are missing to allow us to release it.\n\nhttps://opendev.org/openstack/governance/commit/e90a44648cb2224f9ca5efef6f10ffea963fe6d1\n\nChange-Id: Iafae4acf9fe561277f91ac9c7a2e4e60b9d58c84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/34b136c75dd2a2f1357f153cfbd68d16f2dcd91e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/34b136c75dd2a2f1357f153cfbd68d16f2dcd91e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
