)]}'
{"id":"openstack%2Fproject-config~796414","triplet_id":"openstack%2Fproject-config~master~I361059c6b62ea240b6fef5a61d254959622199d7","project":"openstack/project-config","branch":"master","topic":"ntpdate","hashtags":[],"change_id":"I361059c6b62ea240b6fef5a61d254959622199d7","subject":"infra-package-needs: stub ntp for Fedora","status":"MERGED","created":"2021-06-15 09:45:11.000000000","updated":"2021-06-15 21:45:53.000000000","submitted":"2021-06-15 21:34:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796414-ntpdate","meta_rev_id":"88a81930e5a5891b14163b373c1e46e065d3ee66","_number":796414,"virtual_id_number":796414,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-15 21:34:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-15 20:47:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-15 20:47:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-15 09:55:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d2f7f683fe2a458674fe07976698459358bfa103","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-06-15 09:45:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6fac3e9524076ce703f0080a1dd693433786ea01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 09:55:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b9f50b235ba84119af246e8238876b73 : SUCCESS in 6m 20s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/cadc35b8299b466ca4f9538d37f3447e : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"6c3ab9bdd8a235e967ac0ed3037de360736408d9","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-06-15 20:47:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\ni will monitor the builds and rollout of this","accounts_in_message":[],"_revision_number":1},{"id":"4cf094851c0246ee05dc952d7cfdd6d90353ae4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 20:47:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3f19aaa12b7ca81508c2555fbab255acc0f42924","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 21:34:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4eaf05723ef74001b76a24ec4d9fe6c9 : SUCCESS in 6m 33s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/59b7860d79d8474a8954c2598f89bc07 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"168ba4309492e346470e8f2335059d4a240ad101","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 21:34:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"88a81930e5a5891b14163b373c1e46e065d3ee66","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 21:45:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/91d14bb3cb5d4f1a9d19fe71f2a2e79a : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"20d69256ffa1bccedaebace7e1fc5f09b963b46b","revisions":{"20d69256ffa1bccedaebace7e1fc5f09b963b46b":{"kind":"REWORK","_number":1,"created":"2021-06-15 09:45:11.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/14/796414/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/14/796414/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/796414/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/796414/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/796414/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/14/796414/1"}}},"commit":{"parents":[{"commit":"042a18d2719a667bb1c4224fdc163930678b95e0","subject":"Merge \"Removing openstack-state-management from the bots\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/042a18d2719a667bb1c4224fdc163930678b95e0"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-06-15 09:43:31.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-06-15 09:43:31.000000000","tz":600},"subject":"infra-package-needs: stub ntp for Fedora","message":"infra-package-needs: stub ntp for Fedora\n\nntp/ntpdate isn\u0027t a package on Fedora any more [1].  Make this like\ncentos 8 above and install chrony for time services.\n\n[1] https://fedoraproject.org/wiki/Changes/NtpReplacement\n\nChange-Id: I361059c6b62ea240b6fef5a61d254959622199d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/20d69256ffa1bccedaebace7e1fc5f09b963b46b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/20d69256ffa1bccedaebace7e1fc5f09b963b46b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
