)]}'
{"id":"openstack%2Fproject-config~796590","triplet_id":"openstack%2Fproject-config~master~I715b2c6d464f4e151255a1fe2a28f72f6794cd20","project":"openstack/project-config","branch":"master","topic":"limestone-mirror","hashtags":[],"change_id":"I715b2c6d464f4e151255a1fe2a28f72f6794cd20","subject":"Revert \"Revert \"Revert \"Revert \"Disable limestone due to mirror issues\"\"\"\"","status":"MERGED","created":"2021-06-16 06:45:35.000000000","updated":"2021-06-16 08:26:14.000000000","submitted":"2021-06-16 07:14:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796590-limestone-mirror","meta_rev_id":"8568be54d9c14f2db69e21ed99544db459f662a3","_number":796590,"virtual_id_number":796590,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-16 07:14:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-16 06:47:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-16 06:47:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-16 07:02:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e67448832bf0de0b61ee548576389b5a4854cefe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-06-16 06:45:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b793cb6eb351d12d71073e5a52fb925a4cb9f78","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-06-16 06:47:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nTaking the liberty of approving this.  I think we need limestone to look into this, and out of US working hours that is not practical","accounts_in_message":[],"_revision_number":1},{"id":"c232f95f175e29c5099237f3d902369e95b491a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 07:02:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/21c66a3dcb63448b8b8fa56d8f5fdbdd : SUCCESS in 5m 48s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/1560aa7a31134e3eaec7b50c09f5fc2b : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":1},{"id":"73a636320114d1603288f6e0b12af70f2f2aee2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 07:03:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"03da82d84bc483f169da8546f682f4838847220e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 07:14:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/65bfe3a557fe42df879fcfea512f20b4 : SUCCESS in 6m 01s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/cf030c7244a14801b86e07a29bc8a520 : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":1},{"id":"c4276d01b0aad4869b256c6bc1a5b8ed1eab9c2c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 07:14:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8568be54d9c14f2db69e21ed99544db459f662a3","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 08:26:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/8b022f55c1e14dc7bd37dcb5df25dd42 : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2a9f0c79234b40574fbc76b6115f97df6d46d240","revisions":{"2a9f0c79234b40574fbc76b6115f97df6d46d240":{"kind":"REWORK","_number":1,"created":"2021-06-16 06:45:35.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/90/796590/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/90/796590/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/796590/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/796590/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/796590/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/90/796590/1"}}},"commit":{"parents":[{"commit":"263548c4114b458c1cc7f68b3f1e16b5c770aac9","subject":"Merge \"infra-package-needs: don\u0027t start ntp for Fedora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/263548c4114b458c1cc7f68b3f1e16b5c770aac9"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-06-16 06:44:22.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-06-16 06:45:25.000000000","tz":600},"subject":"Revert \"Revert \"Revert \"Revert \"Disable limestone due to mirror issues\"\"\"\"","message":"Revert \"Revert \"Revert \"Revert \"Disable limestone due to mirror issues\"\"\"\"\n\nUnfortunately the mirror issues persist.  Again it looks likley\nrelated to the attached storage volume and I/O performance,\nparticularly for the http cache.\n\nThis reverts commit 377d797b3fa0e95bb79989b86fde57ea4241190c.\n\nChange-Id: I715b2c6d464f4e151255a1fe2a28f72f6794cd20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2a9f0c79234b40574fbc76b6115f97df6d46d240"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2a9f0c79234b40574fbc76b6115f97df6d46d240"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
